)]}'
{"id":"openstack%2Freleases~832433","triplet_id":"openstack%2Freleases~master~If22df563d5798a8f7c09205754daf3d00f0bdecc","project":"openstack/releases","branch":"master","topic":"yoga-rc1-deadline","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-03-11 10:40:51.000000000","reason":"Change was submitted"},"25701":{"account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"last_update":"2022-03-11 10:40:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If22df563d5798a8f7c09205754daf3d00f0bdecc","subject":"Release Yoga RC1 for tacker","status":"MERGED","created":"2022-03-07 23:22:19.000000000","updated":"2022-03-11 10:40:51.000000000","submitted":"2022-03-11 10:40:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"832433-yoga-rc1-deadline","meta_rev_id":"f840af9e5c95110757dffe03e03352f11bf0e492","_number":832433,"virtual_id_number":832433,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-11 10:40:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"all":[{"value":1,"date":"2022-03-11 09:43:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":2,"date":"2022-03-11 09:59:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2022-03-11 10:06:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-11 10:06:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2022-03-11 10:06:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-03-11 09:42:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-07 23:23:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-07 23:31:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-08 00:21:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2022-03-11 10:06:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-03-11 10:06:50.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"fca4f0f24d6ca7305b2a38224f241df805425ef1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-03-07 23:22:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1907cd144d764a5f41cc0806e3a096fbe27cb84c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 23:31:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a80d6ec6040f4823b1a12719a7a8c43f : SUCCESS in 7m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/767febb5557746e2b7552be45071b98a : SUCCESS in 6m 12s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f27e45c857c842fab8b45f65e47e61b3 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":1},{"id":"e32315b8cf02d82ec47a4e1b09b0bc214b9717ea","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2022-03-08 01:56:16.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ec4e708b0d28dc3e302f2eff836c684309dca1a8","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-03-11 09:29:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6d0ffca4c7c7479487f32575b19687ecfd7bf027","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2022-03-11 09:40:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"14d0e10b9092704af3bcd51b9f9a2f096fd03dd3","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 09:42:53.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"24791f112653541838094f298961b1cdb878fac7","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2022-03-11 09:43:25.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8115d369255dcad2de6d57a2a0fbf674321ce65b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 09:55:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/859ff048bccc4596a6bbac54070b1553 : SUCCESS in 8m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/48f59bde0e4d4846a15651d998e5eaed : SUCCESS in 7m 56s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4b26fc40bd3f45c1a8361a6350873be9 : SUCCESS in 7m 12s","accounts_in_message":[],"_revision_number":2},{"id":"4bfaa9f9c9325ad9242990cffc54e0e7d64c7a26","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-03-11 09:59:30.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"25d985be11b15dc8e5c0f44b80ee3919ce106c39","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-03-11 10:06:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fe215926627c54f88cb12d7efb00b9ec90a27518","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-03-11 10:06:50.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"aa28c1ae0cb9fe1dc422d8ae92e3e0183c260268","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 10:09:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b24d137e05788b70a48f24396258baa35e5baaa5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 10:40:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4603788a7c5b42bf9cb73a32caa5aa07 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/489a946528c84c2bb91fc7c4b674cac5 : SUCCESS in 8m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/81ec4c5735094b08b0b3d4955e9820e0 : SUCCESS in 7m 35s","accounts_in_message":[],"_revision_number":2},{"id":"f840af9e5c95110757dffe03e03352f11bf0e492","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 10:40:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1b56bcac5b28fe06d5b98477a5cb36b4a7509f19","revisions":{"72799349cb07156c344a210a6ab0feb259b79db4":{"kind":"REWORK","_number":1,"created":"2022-03-07 23:22:19.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/33/832433/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/33/832433/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/832433/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/832433/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/832433/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/33/832433/1"}}},"commit":{"parents":[{"commit":"3586c80c936af58827097f80b94140c67a6bf573","subject":"Merge \"Minor release (1.6.0) for validations-common with dedicated stable branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3586c80c936af58827097f80b94140c67a6bf573"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-03-07 23:22:14.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-03-07 23:22:14.000000000","tz":60},"subject":"Release Yoga RC1 for tacker","message":"Release Yoga RC1 for tacker\n\nThis patch proposes an RC1 release for tacker to get any merged\nchanges out prior to the RC1 deadline for Yoga going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Thursday, March 10th.\n\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: If22df563d5798a8f7c09205754daf3d00f0bdecc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/72799349cb07156c344a210a6ab0feb259b79db4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/72799349cb07156c344a210a6ab0feb259b79db4"}]},"branch":"refs/heads/master"},"1b56bcac5b28fe06d5b98477a5cb36b4a7509f19":{"kind":"REWORK","_number":2,"created":"2022-03-11 09:40:45.000000000","uploader":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"ref":"refs/changes/33/832433/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/33/832433/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/832433/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/832433/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/33/832433/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/33/832433/2"}}},"commit":{"parents":[{"commit":"3586c80c936af58827097f80b94140c67a6bf573","subject":"Merge \"Minor release (1.6.0) for validations-common with dedicated stable branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3586c80c936af58827097f80b94140c67a6bf573"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-03-07 23:22:14.000000000","tz":60},"committer":{"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","date":"2022-03-11 09:40:23.000000000","tz":540},"subject":"Release Yoga RC1 for tacker","message":"Release Yoga RC1 for tacker\n\nThis patch proposes an RC1 release for tacker to get any merged\nchanges out prior to the RC1 deadline for Yoga going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Thursday, March 10th.\n\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: If22df563d5798a8f7c09205754daf3d00f0bdecc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1b56bcac5b28fe06d5b98477a5cb36b4a7509f19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1b56bcac5b28fe06d5b98477a5cb36b4a7509f19"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
