)]}'
{"id":"openstack%2Freleases~834042","triplet_id":"openstack%2Freleases~master~I2013576b3e0ca2d86c6990873d5fcbc4a2854dfc","project":"openstack/releases","branch":"master","topic":"puppet-eol","hashtags":[],"change_id":"I2013576b3e0ca2d86c6990873d5fcbc4a2854dfc","subject":"[Puppet OpenStack] Transition stein to EOL","status":"ABANDONED","created":"2022-03-16 15:35:40.000000000","updated":"2022-03-17 06:49:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"aebf582c09cb9965071ecc99a53c20c1dfcf3962","_number":834042,"virtual_id_number":834042,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-03-16 15:54:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-16 15:37:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-16 15:54:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"520beb0920d632064b37403bb0f582a431ada446","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-16 15:35:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0e01d603746d1dc26430c8f2177a56b1a8d944d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 15:54:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d373dff722f9478a8f757ccd5006b69e : SUCCESS in 8m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ca3b301795504515a8b15105cf28cdc8 : SUCCESS in 9m 59s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7a3ab75fb8f642fc9b81dbfea03eabb1 : SUCCESS in 11m 15s","accounts_in_message":[],"_revision_number":1},{"id":"aebf582c09cb9965071ecc99a53c20c1dfcf3962","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-17 06:49:57.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3caee91c906df12d9102ab3500593832fe83b761","revisions":{"3caee91c906df12d9102ab3500593832fe83b761":{"kind":"REWORK","_number":1,"created":"2022-03-16 15:35:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/834042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/42/834042/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/834042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/834042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/834042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/42/834042/1"}}},"commit":{"parents":[{"commit":"38ccc5ce0e7d011b3dddff6f6e113ba7a15ab054","subject":"[Puppet OpenStack] Transition rocky to EOL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/38ccc5ce0e7d011b3dddff6f6e113ba7a15ab054"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 11:58:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-16 15:35:30.000000000","tz":540},"subject":"[Puppet OpenStack] Transition stein to EOL","message":"[Puppet OpenStack] Transition stein to EOL\n\nThis is the last step to retire old stable branches in Puppet\nOpenStack repos and makes stable/stein EOL. The proposal was sent to\nthe mailing list[1] and we haven\u0027t heard any objections.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-October/025589.html\n\nChange-Id: I2013576b3e0ca2d86c6990873d5fcbc4a2854dfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3caee91c906df12d9102ab3500593832fe83b761"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3caee91c906df12d9102ab3500593832fe83b761"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
