)]}'
{"id":"openstack%2Freleases~835205","triplet_id":"openstack%2Freleases~master~Iade172d851b33394a801ca48a608a35e39d01a81","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"Iade172d851b33394a801ca48a608a35e39d01a81","subject":"Release hacking 4.1.1","status":"ABANDONED","created":"2022-03-25 10:59:59.000000000","updated":"2022-08-02 22:18:31.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"13921cbebc84d645593c222aa03188c0dec56c95","_number":835205,"virtual_id_number":835205,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-03-25 11:09:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":-1,"date":"2022-03-25 20:00:54.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-03-25 11:01:08.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-25 11:01:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2d0eede71dded55022c4c42aa8bf346fa705edb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-03-25 10:59:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bfe0a6fc77dfcd0e8da38cdaa58b998e2d409678","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 11:01:08.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"0c62cd8258acc5a39562b47865ebcf40dcefa929","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 11:09:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b12230746064476ba23627a49ed9ce83 : SUCCESS in 8m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a88284a8f8d8406991e16f92f1f4df1c : SUCCESS in 7m 02s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5e49bb205fe346fa97d681ad2e248341 : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":1},{"id":"e4e7c6b8a7616fe62517e7f4b5018cdb33023cd7","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-03-25 20:00:54.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"13921cbebc84d645593c222aa03188c0dec56c95","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-08-02 22:18:31.000000000","message":"Abandoned\n\nin favor of https://review.opendev.org/c/openstack/releases/+/851273","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0d6c4e91dae82b56b226fb53e8d951394b953141","revisions":{"0d6c4e91dae82b56b226fb53e8d951394b953141":{"kind":"REWORK","_number":1,"created":"2022-03-25 10:59:59.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/05/835205/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/05/835205/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/835205/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/835205/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/835205/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/05/835205/1"}}},"commit":{"parents":[{"commit":"3c2190dc2cb01983f91cac6e3760c13f944089e0","subject":"Merge \"Final RC patch for magnum\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3c2190dc2cb01983f91cac6e3760c13f944089e0"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-03-25 10:59:37.000000000","tz":60},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-03-25 10:59:37.000000000","tz":60},"subject":"Release hacking 4.1.1","message":"Release hacking 4.1.1\n\nChange-Id: Iade172d851b33394a801ca48a608a35e39d01a81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0d6c4e91dae82b56b226fb53e8d951394b953141"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0d6c4e91dae82b56b226fb53e8d951394b953141"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
