)]}'
{"id":"openstack%2Freleases~837128","triplet_id":"openstack%2Freleases~master~I5d8da3aa071949d194a369efecf880f042e3762e","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2022-04-11 12:54:49.000000000","reason":"Change was submitted"},"10234":{"account":{"_account_id":10234,"name":"Andrey Pavlov","email":"andrey.mp@gmail.com","username":"andrey-mp"},"last_update":"2022-04-08 14:14:43.000000000","reason":"Andrey Pavlov replied on the change"}},"hashtags":[],"change_id":"I5d8da3aa071949d194a369efecf880f042e3762e","subject":"Create bug release for ec2-api in yoga","status":"MERGED","created":"2022-04-08 13:42:20.000000000","updated":"2022-04-11 12:54:49.000000000","submitted":"2022-04-11 12:54:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837128","meta_rev_id":"6a4512bae065fe66f751fbfd627d24a4c2f5ea9c","_number":837128,"virtual_id_number":837128,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":10234,"name":"Andrey Pavlov","email":"andrey.mp@gmail.com","username":"andrey-mp"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-11 12:54:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2022-04-11 10:43:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2022-04-08 14:14:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10234,"name":"Andrey Pavlov","email":"andrey.mp@gmail.com","username":"andrey-mp"},{"value":2,"date":"2022-04-11 12:43:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-08 15:08:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":10234,"name":"Andrey Pavlov","email":"andrey.mp@gmail.com","username":"andrey-mp"},{"value":1,"date":"2022-04-11 12:43:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":10234,"name":"Andrey Pavlov","email":"andrey.mp@gmail.com","username":"andrey-mp"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-04-08 14:15:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":10234,"name":"Andrey Pavlov","email":"andrey.mp@gmail.com","username":"andrey-mp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-08 13:43:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-08 13:43:48.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":10234,"name":"Andrey Pavlov","email":"andrey.mp@gmail.com","username":"andrey-mp"},"state":"REVIEWER"},{"updated":"2022-04-08 13:50:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-08 15:08:13.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"state":"REVIEWER"},{"updated":"2022-04-11 10:43:10.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2022-04-11 12:43:48.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"e52dd8e684699365dc41c2634ebf4da2e7a4fef5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-04-08 13:42:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f01cd493e887b58634a54f9647e7c4bc76b627c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 13:50:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d708c3f31c1498c9a82b773631b33e5 : SUCCESS in 6m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6d01ad4103b444a98944782dca57a10b : SUCCESS in 4m 49s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/36375d6b70514bbc9e881d0920a405e4 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"7585b1bb99632505aad1c12b72607fdbf8436f81","author":{"_account_id":10234,"name":"Andrey Pavlov","email":"andrey.mp@gmail.com","username":"andrey-mp"},"date":"2022-04-08 14:14:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1ba22c424a350c313ebd6c5b4b423d2730dd306e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 14:15:53.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"02b5b836ba34c9e06f92b0f4e535b9481ee40c4d","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2022-04-08 15:08:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f2e79f1dbfeed9ee4371a2e64101fc505c6128dc","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-04-11 10:43:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b0fb9a46f90290d222070d31ca4ab3af60c06ea3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-11 12:43:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f075f93856bcd8b9b2d08570ec8bcc26da207e56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 12:44:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e2f61d45bc172a871555af593d09d9a1d19daf47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 12:54:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/793d0595c6b649328967ee093fc3f22b : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afe7b9d6e96e499eb28b0ce4d87f8275 : SUCCESS in 7m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8b77d45cc2844c42a463e566d6acc0ca : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":1},{"id":"6a4512bae065fe66f751fbfd627d24a4c2f5ea9c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 12:54:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a8b3f017d0f47a13731d0d72c7e0c3fe41a0c01a","revisions":{"a8b3f017d0f47a13731d0d72c7e0c3fe41a0c01a":{"kind":"REWORK","_number":1,"created":"2022-04-08 13:42:20.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/28/837128/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/28/837128/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/837128/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/837128/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/837128/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/28/837128/1"}}},"commit":{"parents":[{"commit":"dde3365138e83a2bc4119a8694b7eb325bfc9313","subject":"Merge \"Release kolla, kolla-ansible and ansible-collection-kolla Yoga RC1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dde3365138e83a2bc4119a8694b7eb325bfc9313"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-04-08 13:41:15.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2022-04-08 13:41:15.000000000","tz":120},"subject":"Create bug release for ec2-api in yoga","message":"Create bug release for ec2-api in yoga\n\nIt includes a fix required to run with oslo-context from yoga [1].\n\n[1] https://review.opendev.org/c/openstack/ec2-api/+/832547\n\nChange-Id: I5d8da3aa071949d194a369efecf880f042e3762e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a8b3f017d0f47a13731d0d72c7e0c3fe41a0c01a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a8b3f017d0f47a13731d0d72c7e0c3fe41a0c01a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
