)]}'
{"id":"openstack%2Freleases~837995","triplet_id":"openstack%2Freleases~master~I6986e89c07331f08ab344f31401a0e874a23c9c4","project":"openstack/releases","branch":"master","topic":"victoria-em","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-05-05 09:06:15.000000000","reason":"Elod Illes replied on the change"}},"hashtags":[],"change_id":"I6986e89c07331f08ab344f31401a0e874a23c9c4","subject":"Set Victoria status to Extended Maintenance","status":"MERGED","created":"2022-04-14 18:23:05.000000000","updated":"2022-05-05 09:16:37.000000000","submitted":"2022-05-05 09:16:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837995-victoria-em","meta_rev_id":"04842fd18ed933179e968899d1d084e0036659c1","_number":837995,"virtual_id_number":837995,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-05 09:16:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2022-05-05 09:06:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2022-05-05 07:57:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2022-05-05 09:06:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-14 18:48:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-05 07:57:42.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"2fca51687ae9ae2200c1af8074534bd415b43e0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-04-14 18:23:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"39163cc6d7bac819d57c5fccad7c0a5d2753dd61","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-04-14 18:23:35.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"567f810d711625e2a39338068477029ac3417dfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 18:48:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87a116b8701f480b805a315cf295b712 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41f61f19153e445c97e94832946fe827 : SUCCESS in 7m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/db5019d90b7d454aa7e3054744466cd9 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":1},{"id":"dea5fbd98424e3d954fb8b06cab793f5b420affb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-05-04 17:32:28.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"89f5c893d9955b8b03cf7a503b959e798f17c3bd","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-05-04 17:32:49.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"43f4de7785037900ee30bbffa9950d68f0ec3cd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 17:43:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/485ff9829b204e72bfde81a09b58dceb : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1439992d46694e0db9a70a798962261a : SUCCESS in 6m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6a3764ce1a374e20a24bb4853d2e75f9 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":2},{"id":"58a20c26128edaea294bd00a9e87a22f79a68bbf","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-05 07:57:42.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a6af946ca4af31446a6e9099d34418c1a272cfbe","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-05-05 09:06:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6537a4c141b2985e283309628c5909bb88c4c53e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-05-05 09:06:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1eb3d446a1d98bba5769af25beb618c33b736e13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 09:06:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8295e2b1a24634398cbcedd0d134908d7788add5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 09:16:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/312241191dfb46dc8750a0b65103330f : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71912fe40a3f4f42866f382d0432da41 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2992e94413684b18ba9e6e6ce872eb93 : SUCCESS in 7m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0345dcbb40bf41dcb774bfdcbc2e502f : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":2},{"id":"04842fd18ed933179e968899d1d084e0036659c1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 09:16:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4b3c68ef7aeced0b81abfe4974f04520b1567c5e","revisions":{"ec969e531a38eed4bd65ae8c9a82e2c523cda4ff":{"kind":"REWORK","_number":1,"created":"2022-04-14 18:23:05.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/95/837995/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/837995/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/837995/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/837995/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/837995/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/837995/1"}}},"commit":{"parents":[{"commit":"855707f3aef512c89defa78cdfed046451ebdb7f","subject":"Merge \"[manila] Release manila-tempest-plugin 1.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/855707f3aef512c89defa78cdfed046451ebdb7f"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-04-14 18:20:45.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-04-14 18:20:45.000000000","tz":120},"subject":"Set Victoria status to Extended Maintenance","message":"Set Victoria status to Extended Maintenance\n\nProject\u0027s last releases in Victoria have been tagged with victoria-em.\nNow indicate that Victoria is in Extended Maintenance phase also at\nhttps://releases.openstack.org\n\nPlanned transition date is April 27th, 2022.\n\nChange-Id: I6986e89c07331f08ab344f31401a0e874a23c9c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ec969e531a38eed4bd65ae8c9a82e2c523cda4ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ec969e531a38eed4bd65ae8c9a82e2c523cda4ff"}]},"branch":"refs/heads/master"},"4b3c68ef7aeced0b81abfe4974f04520b1567c5e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-05-04 17:32:28.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/95/837995/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/837995/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/837995/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/837995/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/837995/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/837995/2"}}},"commit":{"parents":[{"commit":"445b3c841c47b85f5f9ec80e191e6d6121c7287d","subject":"[OpenStackAnsible] Transition Victoria to EM","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/445b3c841c47b85f5f9ec80e191e6d6121c7287d"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-04-14 18:20:45.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-05-04 17:32:28.000000000","tz":0},"subject":"Set Victoria status to Extended Maintenance","message":"Set Victoria status to Extended Maintenance\n\nProject\u0027s last releases in Victoria have been tagged with victoria-em.\nNow indicate that Victoria is in Extended Maintenance phase also at\nhttps://releases.openstack.org\n\nPlanned transition date is April 27th, 2022.\n\nChange-Id: I6986e89c07331f08ab344f31401a0e874a23c9c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4b3c68ef7aeced0b81abfe4974f04520b1567c5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4b3c68ef7aeced0b81abfe4974f04520b1567c5e"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
