)]}'
{"id":"openstack%2Freleases~838317","triplet_id":"openstack%2Freleases~master~I3245e5cfc8af5b1244a9579f72117686e2855ba7","project":"openstack/releases","branch":"master","topic":"openstacksdk_0.62.0","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-04-28 15:11:14.000000000","reason":"Rodolfo Alonso replied on the change"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2022-05-11 10:33:35.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3245e5cfc8af5b1244a9579f72117686e2855ba7","subject":"Bump openstacksdk to version 0.62.0","status":"ABANDONED","created":"2022-04-18 15:15:56.000000000","updated":"2022-05-11 10:33:35.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"df5a15628990db4186cd0fa933c2eb2bc04faf47","_number":838317,"virtual_id_number":838317,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-04-18 15:24:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":-1,"date":"2022-04-28 13:38:07.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":-1,"date":"2022-04-29 16:41:41.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-18 15:17:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-18 15:20:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2022-04-18 15:24:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1234c82f7b6614a60f651e8eead2a944fe9c4d5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-18 15:15:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae4da62f5a8229a55b8962b9cdc3b1e6f651b39f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 15:24:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2d6bc264fdc4f12923b486be440fdec : SUCCESS in 7m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6299e43641b440558963e7ca0d930844 : SUCCESS in 5m 18s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/37689d2e10ca4366a00fff7884f77c59 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"682ad1b6904ad66608c7ede41397208850a747d1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-28 13:23:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2f3129e856d914d06079388b51f71a008590c975","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-04-28 13:38:07.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"17cae680bbd93905943b815461e7fa03bf919309","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-28 15:11:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"35beccf1e71f9244fd9b5bf6fd21ea3bf58f8cbf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-29 16:41:41.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"df5a15628990db4186cd0fa933c2eb2bc04faf47","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-05-11 10:33:35.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7fdd97720a5f12b8f77ecfcd6e304f36ce6551fc","revisions":{"7fdd97720a5f12b8f77ecfcd6e304f36ce6551fc":{"kind":"REWORK","_number":1,"created":"2022-04-18 15:15:56.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/17/838317/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/17/838317/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/838317/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/838317/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/838317/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/17/838317/1"}}},"commit":{"parents":[{"commit":"ab5dcd10e965d4bc28634ca868df9cb635660806","subject":"Merge \"Add release note links for trailing projects for Yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ab5dcd10e965d4bc28634ca868df9cb635660806"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-04-13 16:06:04.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-04-13 16:06:04.000000000","tz":0},"subject":"Bump openstacksdk to version 0.62.0","message":"Bump openstacksdk to version 0.62.0\n\nIncluded commits:\n6b2f555a image: Add \"id\" filter for images\n975cabbd Add QoS rule type filtering keys\n53e81c16 Add Python3 zed unit tests\n194f3545 Update master for stable/yoga\nbdd87c70 Identity: Add support for system role assignment\n4c1ffee6 Indentation of the docstrings\n44c55156 network: Fix update of network provider\n55eff088 Correct documentation about Load Balancer API\nbcd5916c Improve compute flavor handling\ndd892d7b Use empty read_acl for swift container\n26a5d10d Restore get_compute_limits backward compatibility\n1b479f0b Revive legacy job\n7569cd9a Skip tests when needed extensions are disabled\nfe3d36f0 Cloud / Utils service - reindentation of the docstrings\n827736d3 Cloud / Network service - reindentation of the docstrings\n6650374b Cloud / Identity service - reindentation of the docstrings\n3244abd0 tests: Centralize configuration of default flavor, image\n2ee2df6b Skip qos-pps-minimum tests if extension is missing\n11d89450 doc: Update testing documentation\n05a83963 doc: Remove references to \u0027examples\u0027 test env\n1f1aa4fa Add query parameters to local ip\n2d69560e Cloud / Compute service - reindentation of the docstrings\n441b7d0c Cloud / Dns service - reindentation of the docstrings\n63aa7c00 Cloud / Coe service - reindentation of the docstrings\n0ce0fddc Cloud / Baremetal service - reindentation of the docstrings\n582e301b Cloud / Security group service - reindentation of the\ndocstrings\n560e1f91 Cloud / Normalize service - reindentation of the docstrings\ndd9b95c1 Cloud / Floating service - reindentation of the docstrings\nafb49692 fix creation of protected image\n5e969fb4 Revert \"Add \"security_group_ids\" to Port\u0027s query parameters\"\n3c8826d0 Get rid of normalization in compute CL\ncadee6fa Functional tests for vpn ike policy resource\n3e84351c Add QoS min pps rule object and CRUD operations\n4f54021e Get rid of normalization in network CL\ned86ffbf Get rid of normalization in further CL services\ne910e6ea Get rid of normalization in identity CL\nf34ea0d1 Get rid of normalization in orchestration CL\n6c905301 Get rid of normalization for image service\nfc2d56fa Object Store - reindentation of the docstrings\nee24cbff Switch quota_set cloud layer BS methods to proxy\nefdc8c2a Switch unittests for BS to use v3\n82f09b4e Reindentation of the docstrings for baremetal service\n4831e22a Switch quota methods of cloud layer to proxy\n29c87246 Switch create_server cloud method to proxy\nc0a9b64e Switch delete_server cloud method to rely on proxy\n9d458860 tests: Handle overridden \u0027verify_delete\u0027 (kw)args\nf4dafb3a Switch update_server cloud layer to proxy\na92ee7ae Switch rebuild_server cloud method to rely on proxy\nd7098876 Fix misuse of assertTrue\nb2bea061 Identity service - reunite :class: links on single lines\n6775fb80 Block storage - reunite :class: links on single lines\ne94fda9e Switch cloud.compute.get_server_by_id to use proxy\n104a361a Add revert share to snapshot to shared file system.\n7c003e6a Shared file system - reunite :class: links on single lines\nda717351 Instance HA service - reunite :class: links on single lines\na1263076 Load balancer - reindentation of the docstrings\n831f7bd5 Network service - reunite :class: links on single lines\n9290eeea Orchestration service - reunite class links (:class:) on one\nline\n02c175db Compute - reindentation of the docstrings\n4f8fae5c Add Neutron Local IP CRUD\n8b09b8f1 Splits class `TestNetworkProxy`\ncc00aafd Vpn ike policy resource\n6fdc97b7 compute: Add support for server lock reason\n6487cad2 Add \"check_limit\" to network Quota class\nf5b60d2b Add Neutron Local IP CRUD\n94c6131c trivial: Correct docstrings for \u0027delete\u0027 proxy calls\n353a0ffe compute: Server group rules are dicts, not lists of dicts\n3692438f compute: Add support for microversion 2.89\n41893385 Fix functional test for user message\nee75531e Adds \"test_update_zone\" test case\nb48d0b8c Database service - reindentation of the docstrings\n15da4d22 Image service - reindentation of the docstrings\naa58c884 Orchestration- reindentation of the docstrings\n42468112 Keep creating keystone admin endpoint for heat\nef1bee16 Switch compute limits cloud to proxy\n\nChange-Id: I3245e5cfc8af5b1244a9579f72117686e2855ba7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7fdd97720a5f12b8f77ecfcd6e304f36ce6551fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7fdd97720a5f12b8f77ecfcd6e304f36ce6551fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
