)]}'
{"id":"openstack%2Freleases~840828","triplet_id":"openstack%2Freleases~master~Id7981e6b0b04cf202094a30bd1818f03f77ff01c","project":"openstack/releases","branch":"master","topic":"neutron-lib_2.21.0","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-05-10 08:27:57.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-05-10 08:27:57.000000000","reason":"Change was submitted"},"841":{"account":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"last_update":"2022-05-10 08:27:57.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-05-09 15:37:53.000000000","reason":"Lajos Katona replied on the change"},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2022-05-10 08:27:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id7981e6b0b04cf202094a30bd1818f03f77ff01c","subject":"Bump neutron-lib to 2.21.0","status":"MERGED","created":"2022-05-06 08:22:20.000000000","updated":"2022-05-10 08:27:57.000000000","submitted":"2022-05-10 08:27:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840828-neutron-lib_2.21.0","meta_rev_id":"c97c48159ad7d4f217c7a5d732482ce926e761ce","_number":840828,"virtual_id_number":840828,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-10 08:27:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2022-05-09 16:30:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2022-05-09 15:37:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2022-05-10 08:20:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2022-05-10 08:20:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-05-09 15:38:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-06 08:23:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-06 08:23:30.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-05-06 08:23:30.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-05-06 08:23:30.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-05-06 08:29:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-06 11:20:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2022-05-09 13:56:50.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2022-05-09 16:30:28.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2022-05-10 08:20:33.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"2a48f63482f8acb45eb61672d37da39b2aa92d36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-05-06 08:22:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8b0b67e5f43494f71a871c1079a911b21393612","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 08:29:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a797de95cd7245919c335c4dd1297c8a : SUCCESS in 6m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1dddc9049fb44f1a8f455bbdc122e6ee : FAILURE in 5m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c9f5c1fea1c14a619ec44a4933ade1d6 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":1},{"id":"e384a4c48224fbf4f71c1f207883e8daf24ec2b7","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-05-06 09:53:03.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"dfbca467ff7534ecd0ad6157abf73d5f22900275","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-05-06 14:21:54.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7bb2007002bc09763f03778cee8675e528d2de6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-05-06 14:24:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3e1ea130b8b658f46c6e7345011f5075ede93195","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 14:32:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/878281de61ed49bc98c81c4eea90847a : SUCCESS in 6m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1e22e932357b48579e2f7d677904ff01 : FAILURE in 5m 31s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b7eff9d2ea3a4d91b69bdb8e63549208 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":2},{"id":"6c7ad267715d058de1830ffb668d4b1a14aebe12","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-05-09 13:56:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"91c55aa0b87361fc558b30f390fc536b604253c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 14:05:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8423860378c647e78461cc9c650d2e8a : SUCCESS in 5m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/45fcbdad3a6647fb8a164aab6c8d8597 : SUCCESS in 4m 50s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b2ef6cb8798943fbad560b7358192cad : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":2},{"id":"6078c6fc2f35938aa6fb257ba79c1d70463ed0d3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-05-09 15:37:53.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"df269995146b9b24e8ba5d8d39f8daca468679cf","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 15:38:53.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"fe3ad8337869d22bee5018003fef0688c0241dca","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-05-09 16:30:28.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"acb6d97f73d8278ead57f87d8694dd4c64f50188","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-10 08:20:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1f9c4371fb3516a01889358a70d0dd52ae216f1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 08:20:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4c7ae881be5d058b796d59277076b83548cd018b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 08:27:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/598cc16b5cef4d8397b8cb983f986017 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/156c8c8a34ec495e898e7fbb497e2486 : SUCCESS in 6m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6df1f7f62f1f43ae90bf2f6e31314dc6 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":2},{"id":"c97c48159ad7d4f217c7a5d732482ce926e761ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 08:27:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8130daea531356b2daaefe7016ca132a7b68ad8e","revisions":{"21aea26f7d620ede550bf2cdcbef23d39d070b85":{"kind":"REWORK","_number":1,"created":"2022-05-06 08:22:20.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/28/840828/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/28/840828/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/840828/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/840828/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/840828/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/28/840828/1"}}},"commit":{"parents":[{"commit":"f7beb542abb256a1e241c45b3a389ecab911583e","subject":"Merge \"Replace old sdist and wheel build command in validate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f7beb542abb256a1e241c45b3a389ecab911583e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-05-05 18:32:56.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-05-05 18:32:56.000000000","tz":0},"subject":"Bump neutron-lib to 2.21.0","message":"Bump neutron-lib to 2.21.0\n\n$ git log --oneline --no-merges 2.20.0...daaa1e7\ndaaa1e7 (HEAD, origin/master) Drop lower-constraints.txt and its testing\n2dfe3c3 Remove unimplemented api-def: BFD monitors\nef1e8ab Refactor session \"is_active\" handling for sqlalchemy-20\n8448eab Add \"neutron-tempest-plugin-jobs\" to CI\na5b9390 Remove unimplemented api-def: multiple-external-gateways\n75a5fbc Don\u0027t translate logs\n652741d doc: add local-ip api-ref to index\n7ae8e74 Doc: Add router gateway IP QoS to api-ref\nda2baf3 Add oslo_policy.InvalidScope exception to the api faults map\nad835ad Rehome missing ovs constants into neutron-lib\na3ee9e0 Remove unused code\nb5f0a26 Add port-mac-address-override shim extension\n74d3765 Add Python3 zed unit tests\n5ef0d26 Add NumaAffinityPoliciesEnumField class unit tests\n1dde6c2 Add custom range parameters on port_range validator\n096fb92 Rehome ``NumaAffinityPoliciesEnumField``\na8abe9d [test]Remove the validation of the tenant property\ne2f05ff Support binding floating ip to smart-nic port\n38d6345 Update master for stable/yoga\nf765e3c Rehome exception ``L3ExtensionException``\n\nChange-Id: Id7981e6b0b04cf202094a30bd1818f03f77ff01c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/21aea26f7d620ede550bf2cdcbef23d39d070b85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/21aea26f7d620ede550bf2cdcbef23d39d070b85"}]},"branch":"refs/heads/master"},"8130daea531356b2daaefe7016ca132a7b68ad8e":{"kind":"REWORK","_number":2,"created":"2022-05-06 14:24:45.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/28/840828/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/28/840828/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/840828/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/840828/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/28/840828/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/28/840828/2"}}},"commit":{"parents":[{"commit":"f7beb542abb256a1e241c45b3a389ecab911583e","subject":"Merge \"Replace old sdist and wheel build command in validate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f7beb542abb256a1e241c45b3a389ecab911583e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-05-05 18:32:56.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-05-06 01:06:23.000000000","tz":0},"subject":"Bump neutron-lib to 2.21.0","message":"Bump neutron-lib to 2.21.0\n\n$ git log --oneline --no-merges 2.20.0...d2b395f\nd2b395f (HEAD, origin/master) fix pep8 error\ndaaa1e7 Drop lower-constraints.txt and its testing\n2dfe3c3 Remove unimplemented api-def: BFD monitors\nef1e8ab Refactor session \"is_active\" handling for sqlalchemy-20\n8448eab Add \"neutron-tempest-plugin-jobs\" to CI\na5b9390 Remove unimplemented api-def: multiple-external-gateways\n75a5fbc Don\u0027t translate logs\n652741d doc: add local-ip api-ref to index\n7ae8e74 Doc: Add router gateway IP QoS to api-ref\nda2baf3 Add oslo_policy.InvalidScope exception to the api faults map\nad835ad Rehome missing ovs constants into neutron-lib\na3ee9e0 Remove unused code\nb5f0a26 Add port-mac-address-override shim extension\n74d3765 Add Python3 zed unit tests\n5ef0d26 Add NumaAffinityPoliciesEnumField class unit tests\n1dde6c2 Add custom range parameters on port_range validator\n096fb92 Rehome ``NumaAffinityPoliciesEnumField``\na8abe9d [test]Remove the validation of the tenant property\ne2f05ff Support binding floating ip to smart-nic port\n38d6345 Update master for stable/yoga\nf765e3c Rehome exception ``L3ExtensionException``\n\nChange-Id: Id7981e6b0b04cf202094a30bd1818f03f77ff01c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8130daea531356b2daaefe7016ca132a7b68ad8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8130daea531356b2daaefe7016ca132a7b68ad8e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
