)]}'
{"id":"openstack%2Freleases~841843","triplet_id":"openstack%2Freleases~master~I3cdcf9f320a46fd45ba494404af7feeba6507727","project":"openstack/releases","branch":"master","topic":"zed-milestone-1","attention_set":{},"removed_from_attention_set":{"15895":{"account":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"last_update":"2022-05-20 19:26:52.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-05-20 19:26:52.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-05-20 19:26:52.000000000","reason":"Change was submitted"},"8731":{"account":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"last_update":"2022-05-20 19:26:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3cdcf9f320a46fd45ba494404af7feeba6507727","subject":"Release mistral-extra for Zed-1 milestone","status":"MERGED","created":"2022-05-16 07:51:56.000000000","updated":"2022-05-20 19:26:52.000000000","submitted":"2022-05-20 19:26:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841843-zed-milestone-1","meta_rev_id":"16ca27e5da6b6a4494832f2d9a93faca7f8336d8","_number":841843,"virtual_id_number":841843,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-20 19:26:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2022-05-20 19:14:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"date":"2022-05-17 14:15:03.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2022-05-20 19:14:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"}],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-16 07:52:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-16 08:01:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-16 08:23:56.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2022-05-16 11:21:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2022-05-16 11:21:25.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2022-05-20 19:14:33.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"cfcd2f441bd9442e0d03a24c7ef7f91a254c2b4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-16 07:51:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49b463d40231b8739401b861333a6ced6edbe046","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 08:01:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c857fb286e9419f803d7a71c273353a : SUCCESS in 8m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0f29eada63d54590841261dffaedc053 : SUCCESS in 5m 51s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f5fec9a9056c4e7baacd2879bd1a17d5 : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":1},{"id":"a01dba6ce0d5592a88806c7749f8b1b0d90d73c1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-16 08:23:56.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8150c0f2122fbbf0cbd4d1f9d15d5bf1a9fe6d3d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-16 10:13:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cb78e409bcbdf9f2074a64a9bd3b19167d342626","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-16 10:16:10.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e75b6ff9d6b13ae997a75327e276dc53171ff918","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 10:27:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c0bdd9c96cd49868b5670bbb437c02e : SUCCESS in 7m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/28b9b33e69f0477bbc07f74a590e53c6 : SUCCESS in 5m 28s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ea1a1e4400334984aafe4d51fe48f921 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":2},{"id":"e1f2a40583a670d940d316572eb1471c04250854","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-16 12:21:49.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f5ce74ad8765bce9fde8f6955e877131b512cb23","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-16 13:00:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5bc593c508b40a8502d4d2a0ab87968c35e923b8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-17 14:15:03.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"40da4e55777fc4a56477f655deddc6410b67dd53","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-18 09:20:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b56a93a33e322d63761ba066ab09db672c193d5a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-05-20 19:14:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"85ddeb5cf893a89fa4ffb814212e466711cc8173","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 19:15:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"72618ffd2127d727a7c08ceca04f0dbe6d2b6d3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 19:26:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31d7191d9c94403c8da9a7caa02358d5 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/475e4e03fb60456ca8fae32d65d07f37 : SUCCESS in 7m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5f3233172e464301bf11160b7cf1ebf4 : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":2},{"id":"16ca27e5da6b6a4494832f2d9a93faca7f8336d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 19:26:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d9086c38386b316de7e6874e598874960bc57499","revisions":{"3eac70fd4ed731b4b36bb49735b05b4253780971":{"kind":"REWORK","_number":1,"created":"2022-05-16 07:51:56.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/43/841843/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/43/841843/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/841843/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/841843/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/841843/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/43/841843/1"}}},"commit":{"parents":[{"commit":"c9d01acbb6ee0b5ceaafd387932089df13a3f434","subject":"Merge \"[Puppet OpenStack] Transition stein to EOL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c9d01acbb6ee0b5ceaafd387932089df13a3f434"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-16 07:51:49.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-16 07:51:49.000000000","tz":120},"subject":"Release mistral-extra for Yoga-1 milestone","message":"Release mistral-extra for Yoga-1 milestone\n\nThis is a library release for mistral-extra for the Yoga-1\nmilestone. This repository contains commits that have not been releases\nfor this cycle yet.\n\nIf the team is ready to process with this release, please leave a +1\nto indicate we should go ahead with the release.\n\nIf the team needs more time for things about to merge, or if there is\nsome other reason a release should not be done at this time, please\nleave a -1 with a comment indicating the status. Then update the patch\nwith the new commit hash to use once the team is ready to do the\nrelease.\n\n$ git log --oneline --no-merges 11.2.0..d8a8f99\nd8a8f99 Remove support for unmaintained Glare\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I3cdcf9f320a46fd45ba494404af7feeba6507727\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3eac70fd4ed731b4b36bb49735b05b4253780971"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3eac70fd4ed731b4b36bb49735b05b4253780971"}]},"branch":"refs/heads/master"},"d9086c38386b316de7e6874e598874960bc57499":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-05-16 10:16:10.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/43/841843/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/43/841843/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/841843/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/841843/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/43/841843/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/43/841843/2"}}},"commit":{"parents":[{"commit":"c9d01acbb6ee0b5ceaafd387932089df13a3f434","subject":"Merge \"[Puppet OpenStack] Transition stein to EOL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c9d01acbb6ee0b5ceaafd387932089df13a3f434"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-16 07:51:49.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2022-05-16 10:16:10.000000000","tz":0},"subject":"Release mistral-extra for Zed-1 milestone","message":"Release mistral-extra for Zed-1 milestone\n\nThis is a library release for mistral-extra for the Zed-1\nmilestone. This repository contains commits that have not been releases\nfor this cycle yet.\n\nIf the team is ready to process with this release, please leave a +1\nto indicate we should go ahead with the release.\n\nIf the team needs more time for things about to merge, or if there is\nsome other reason a release should not be done at this time, please\nleave a -1 with a comment indicating the status. Then update the patch\nwith the new commit hash to use once the team is ready to do the\nrelease.\n\n$ git log --oneline --no-merges 11.2.0..d8a8f99\nd8a8f99 Remove support for unmaintained Glare\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I3cdcf9f320a46fd45ba494404af7feeba6507727\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d9086c38386b316de7e6874e598874960bc57499"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d9086c38386b316de7e6874e598874960bc57499"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
