)]}'
{"id":"openstack%2Freleases~841844","triplet_id":"openstack%2Freleases~master~Ie3ea1b21df8b20ca357e0457d39e9a411b27e784","project":"openstack/releases","branch":"master","topic":"zed-milestone-1","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-05-20 08:22:30.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-05-19 11:30:42.000000000","reason":"Elod Illes replied on the change"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-05-20 08:11:45.000000000","reason":"Hervé Beraud replied on the change"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2022-05-19 06:37:32.000000000","reason":"Artem Goncharov replied on the change"}},"hashtags":[],"change_id":"Ie3ea1b21df8b20ca357e0457d39e9a411b27e784","subject":"Release openstacksdk for Zed-1 milestone","status":"MERGED","created":"2022-05-16 07:55:53.000000000","updated":"2022-05-20 08:22:30.000000000","submitted":"2022-05-20 08:22:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"841844-zed-milestone-1","meta_rev_id":"6bb283e469f5032b969e5e8d9f57ca41f0df38f8","_number":841844,"virtual_id_number":841844,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-20 08:22:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2022-05-19 11:30:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2022-05-20 08:11:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-19 10:11:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2022-05-20 08:11:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-05-19 10:12:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-16 07:57:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-16 08:09:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-16 11:20:44.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2022-05-18 17:49:40.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2022-05-19 07:42:00.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"c92116d2b886d3d05ff58bcec0279cb928f43ba1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-16 07:55:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1df569d2cedaf28ee6f052d7f7684114e3ee6f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 08:09:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e87576e56a64774a171773b29bd1060 : SUCCESS in 8m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1d7ea9b1f5ab483c9c409578d3b081ac : SUCCESS in 5m 34s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f8e54d7d8db74a8cadd2cb40afc6cc59 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"a4861fd8b598f6436f4f4644ff930bb94f5ff121","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-16 10:20:04.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"54e7a230b6b0fa3b89a0458fae41b5ffc57eeea3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 10:29:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb20813a73bc4a7282f5fbe45a1a7fca : SUCCESS in 6m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c07ed961a4024c90853f5e55d2f51347 : SUCCESS in 5m 50s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e746f495445943148cb01ac198ff4db8 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":2},{"id":"f22eda8b2cbfaf941d2244360f8f1f62fa09de91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-05-18 15:17:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ed6694669a31ac9f3836944ac5ba65fdbd87fb6c","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 15:18:46.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"9bf3ffd6a77e4f56b5b0f8eaf5e9f39b48026c2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 15:30:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66f18c400cb74f7f83442af711a52788 : SUCCESS in 7m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b8dfc223a42f46048353ec329590fee7 : FAILURE in 6m 27s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/fbf2005be3944b56b95b938727f9ae0d : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":3},{"id":"d6bb95b4c2e1fb8fc95d10d61f08f8c1b8a9db39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-05-18 16:33:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"048747776474a79179637ce2ab18ca769af6cf65","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 16:34:54.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"ea9a6179fc3cf2e9e0f69fea833fe5307b243c6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 16:44:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bee420651afe4fefafdb14d418d25dd8 : SUCCESS in 8m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dfd821e776f44a7c9e8e4726c376b098 : FAILURE in 5m 50s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/87aaed5c658b4708ba837aa2c18da111 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":4},{"id":"40211dbbc6aadd386a9c54638c36c655d41797f4","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-05-18 17:49:40.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e29fdaf03610a295641363fe8db39d0792140783","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-05-18 18:32:00.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ecb434f8df97a6ed827312cc0808966fadf79221","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 18:44:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03839d2353e040f4ac9d106c92e7433a : SUCCESS in 7m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/145e7260601f4ae4a1eef978fc523a25 : SUCCESS in 5m 54s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/121be7cd193a4d2eb6c4e4bab68e39ec : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":5},{"id":"49f5f0c08ec86efe23200d55268a206cd5cc95af","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-05-19 06:37:32.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2681fbee670cf9b647fc3d7117266720be5996c6","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 06:38:32.000000000","message":"Patch Set 5: PTL-Approved+1","accounts_in_message":[],"_revision_number":5},{"id":"f666c8598536e643cce68c45940f16bda09e8224","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-19 07:25:40.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"0f92348eb546ac05823df5f0bcb33262fcbfe4ec","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-05-19 07:42:00.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4df6857645b286cbe0878c2958c0776c49db4125","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-05-19 09:53:05.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4cf80e3b993e743c0508f61bed3d038383e0281f","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-05-19 10:11:53.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"0bbfdd7f81453565242736c6ebaa9645e97dfdcd","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 10:12:56.000000000","message":"Patch Set 6: PTL-Approved+1","accounts_in_message":[],"_revision_number":6},{"id":"429da2943f625baca512462fa392090dea05ca7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 10:34:16.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5ea6bd89eff44d7b9a5c4d9f767793e : SUCCESS in 6m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a92ca9b390bf4ef9a60ab6f61f361cd1 : SUCCESS in 6m 08s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5ad40a8429c3400883fd808cba6e1d42 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":6},{"id":"6652cd81f5a7c4f5ecaa23f1be4c844ee8b52cc8","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-05-19 11:30:42.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"dd15bd878d63ac198749b384c17efee07bd4e468","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-20 08:11:45.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"0b20fb5b0e25ee71b2f0e2824e8c0a75319dd91d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 08:12:12.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"6766af10a211e0a73fd80f7bd45d64078c08506c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 08:22:30.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb165f990bf949a69e9c2f2d53ead695 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c122313308cc4672b7268a760b22f7c2 : SUCCESS in 7m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1788f447ae4d49668fcba4bcaae4ca24 : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":6},{"id":"6bb283e469f5032b969e5e8d9f57ca41f0df38f8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 08:22:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"7934e97d79437c4297645cdc49909205fd42397a","revisions":{"b05f1110bd8e999c055f354d5c5eee3053d2dcaf":{"kind":"REWORK","_number":1,"created":"2022-05-16 07:55:53.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/44/841844/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/841844/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/841844/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/841844/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/841844/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/841844/1"}}},"commit":{"parents":[{"commit":"c9d01acbb6ee0b5ceaafd387932089df13a3f434","subject":"Merge \"[Puppet OpenStack] Transition stein to EOL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c9d01acbb6ee0b5ceaafd387932089df13a3f434"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-16 07:55:45.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-16 07:55:45.000000000","tz":120},"subject":"Release openstacksdk for Yoga-1 milestone","message":"Release openstacksdk for Yoga-1 milestone\n\nThis is a library release for openstacksdk for the Yoga-1\nmilestone. This repository contains commits that have not been releases\nfor this cycle yet.\n\nIf the team is ready to process with this release, please leave a +1\nto indicate we should go ahead with the release.\n\nIf the team needs more time for things about to merge, or if there is\nsome other reason a release should not be done at this time, please\nleave a -1 with a comment indicating the status. Then update the patch\nwith the new commit hash to use once the team is ready to do the\nrelease.\n\n$ git log --oneline --no-merges 0.61.0..652faab5\n360d517a Improve StatsD metric precision\n1e50ea50 Add R1 summary release note\n6b2f555a image: Add \"id\" filter for images\n46db8d3b Allow to filter endpoints by region_id\n975cabbd Add QoS rule type filtering keys\nbdccc6b7 fix: improperly encoded object names\n358ddcd3 Run nodepool job\n53e81c16 Add Python3 zed unit tests\n194f3545 Update master for stable/yoga\nbdd87c70 Identity: Add support for system role assignment\n4c1ffee6 Indentation of the docstrings\n44c55156 network: Fix update of network provider\n55eff088 Correct documentation about Load Balancer API\nbcd5916c Improve compute flavor handling\ndd892d7b Use empty read_acl for swift container\n26a5d10d Restore get_compute_limits backward compatibility\n1b479f0b Revive legacy job\n7569cd9a Skip tests when needed extensions are disabled\nfe3d36f0 Cloud / Utils service - reindentation of the docstrings\n827736d3 Cloud / Network service - reindentation of the docstrings\n6650374b Cloud / Identity service - reindentation of the docstrings\n3244abd0 tests: Centralize configuration of default flavor, image\n2ee2df6b Skip qos-pps-minimum tests if extension is missing\n11d89450 doc: Update testing documentation\n05a83963 doc: Remove references to \u0027examples\u0027 test env\n1f1aa4fa Add query parameters to local ip\n2d69560e Cloud / Compute service - reindentation of the docstrings\n441b7d0c Cloud / Dns service - reindentation of the docstrings\n63aa7c00 Cloud / Coe service - reindentation of the docstrings\n0ce0fddc Cloud / Baremetal service - reindentation of the docstrings\n582e301b Cloud / Security group service - reindentation of the docstrings\n560e1f91 Cloud / Normalize service - reindentation of the docstrings\ndd9b95c1 Cloud / Floating service - reindentation of the docstrings\nafb49692 fix creation of protected image\n5e969fb4 Revert \"Add \"security_group_ids\" to Port\u0027s query parameters\"\n3c8826d0 Get rid of normalization in compute CL\ncadee6fa Functional tests for vpn ike policy resource\n3e84351c Add QoS min pps rule object and CRUD operations\n4f54021e Get rid of normalization in network CL\ned86ffbf Get rid of normalization in further CL services\ne910e6ea Get rid of normalization in identity CL\nf34ea0d1 Get rid of normalization in orchestration CL\n6c905301 Get rid of normalization for image service\nfc2d56fa Object Store - reindentation of the docstrings\nee24cbff Switch quota_set cloud layer BS methods to proxy\nefdc8c2a Switch unittests for BS to use v3\n82f09b4e Reindentation of the docstrings for baremetal service\n4831e22a Switch quota methods of cloud layer to proxy\n29c87246 Switch create_server cloud method to proxy\nc0a9b64e Switch delete_server cloud method to rely on proxy\n9d458860 tests: Handle overridden \u0027verify_delete\u0027 (kw)args\nf4dafb3a Switch update_server cloud layer to proxy\na92ee7ae Switch rebuild_server cloud method to rely on proxy\nd7098876 Fix misuse of assertTrue\nb2bea061 Identity service - reunite :class: links on single lines\n6775fb80 Block storage - reunite :class: links on single lines\ne94fda9e Switch cloud.compute.get_server_by_id to use proxy\n104a361a Add revert share to snapshot to shared file system.\n7c003e6a Shared file system - reunite :class: links on single lines\nda717351 Instance HA service - reunite :class: links on single lines\na1263076 Load balancer - reindentation of the docstrings\n831f7bd5 Network service - reunite :class: links on single lines\n9290eeea Orchestration service - reunite class links (:class:) on one line\n02c175db Compute - reindentation of the docstrings\nf5b60d2b Add Neutron Local IP CRUD\n94c6131c trivial: Correct docstrings for \u0027delete\u0027 proxy calls\n353a0ffe compute: Server group rules are dicts, not lists of dicts\n41893385 Fix functional test for user message\nb48d0b8c Database service - reindentation of the docstrings\n15da4d22 Image service - reindentation of the docstrings\naa58c884 Orchestration- reindentation of the docstrings\n42468112 Keep creating keystone admin endpoint for heat\nef1bee16 Switch compute limits cloud to proxy\n31c85038 Shared File System - reindentation of the docstrings\n72c223b5 Add share snapshot to shared file system.\nef587c9a compute: Add support for microversion 2.89\n770a3c3e compute: Add support for migrations API\n6d1321f5 compute: Add support for server migrations API\n6c96faa7 compute: Regroup proxy methods\nbf267b0e DNS - reindentation of the docstrings\ne574e4dd Clustering - reindentation of the docstrings\n0fc28b39 Network - reindentation of the docstrings\nb47652cf trivial: Typo fixes\n86c85103 Remove \u0027__unicode__\u0027 helper\n58270810 exception: Correct argument\n95640856 Add support for updated_at field for volume objects\n894b5450 Block storage - reindentation of the docstrings\n9d143ed8 Reindentation of the docstrings\n4581db6c Reindentation of the docstrings\nc746567a Reindentation of the docstrings\nef1b622b Reindentation of the docstrings\n322c72ba Reindentation of the docstrings\n4ad2d0d2 Reindentation of the docstrings\n388a463a Reindentation of the docstrings\n45f400dc resource: Rewrap function signatures\n93a8258e resource: Remove deprecated \u0027allow_get\u0027 attribute\n07eb09f6 Rely on proxy for compute.flavor cloud layer operations\n15a8fc4c Support description in sg-rule creation\n16e557f4 Switch Swift cloud layer to proxy\nfde91622 Splits class `TestSharedFileSystemProxy`\nd3b614ad Splits class `TestPlacementProxy`\n6bdde3d1 Splits class `TestOrchestrationProxy`\n93f69e42 Splits class `TestNetworkProxy`\n81776e61 Add missing headers in object store\n56068af2 Splits class `TestMessageProxy`\n45995c88 Splits class `TestKeyManagerProxy`\nf5f0f9f1 Splits class `TestInstanceHaProxy`\n50043daf Splits the `TestBaremetalProxy` class\n290a6cf5 Add limit resource to shared file system\nbba9c5c5 Fix setting initial object/container metadata.\n6a2420db Splits the `TestAcceleratorProxy` class\na2b7c540 Optimize code by moving misc functions to utils\n7f87d2ee Start splitting the `TestImageProxy` class\n4914d8d7 Add user message to shared file system\n67c8c191 Replace the \u0027try except\u0027 block\nd7d6464c Add user group assignment support in identity\n21795acf Add a manila functional check job\nd62a4e60 Fix key generation for caching\nf7f9446d Add \"security_group_ids\" to Port\u0027s query parameters\n88e520f5 Adds storage pools to shared file system\na8c3cb29 Extend has_version function to accept version parameter\n1909df1f Vpn ike policy resource\n3d6aff34 Add common metadata mixin to block storage and clustering\nc1c8dfb7 Make metadata a common mixin\nf3df2095 Introduce QuotaSet in block storage service\nb7ebc731 Introduce QuotaSet in the compute service\n94cb864d Introduce common class for QuotaSet\nfae7ab06 Fix import order in block_storage tests\n2d56b681 Fix import order in cloud tests\nc6494db0 Move tag mixin into the common\nb59cda5c Fix import order in clustering tests\nadcf98c2 Fix import order in orchestration tests\nb86d3a8c Fix import order in identity tests\n8a1d255c Switch identity roles in the cloud layer to proxy\n95cec287 Fix identity role management\n533ba93a Switch identiy.endpoints in the cloud layer\ndf5f987c Add share resource to shared file system\n9a038d8c Switch identity.service in cloud layer to proxy\ne192f215 Switch identity.domains in cloud layer\n54f910f2 Switch identity groups and users in cloud layer\n74fa9871 Add possibility to create subprojects in cloud layer\n72aa4531 Switch project management in cloud to proxy\nccc6478c Temporarily disable nodepool job\nd6c51c53 Add cleanup function for DNS resource\nf5c9653a Switch BS volume_attachment operations in cloud layer\n485b53bd Switch BS backup and snapshot methods in cloud layer to proxy\n734ae40f Add BS snapshot and backup actions\n118cf735 Switch BS type access operations in cloud layer\n8579b095 Add BS type access into proxy/resource layer\nd235b141 Switch block_storage.volume operations in cloud layer\nf9183cf4 Add block_storage.volume actions\nd1c9a96e Copy find_volume to block_storage.v2 proxy\na3582786 Fix active status for block storage\ne6fed6d2 Add scheduler hints on the block_storage.volume resource\n62781ae2 Sort block storage properties alphabetically\nc6177a47 Drop cloud layer methods for Senlin\n7cee92d5 Replace deprecated inspect.getargspec call\n28875355 Revert \"Update TOX_CONSTRAINTS_FILE for feature/r1\"\n698ac14a Switch object_store cloud functions to proxy layer\n70a06d99 Stop sending tenant_id to Neutron\n6db391f6 Switch networking function in cloud layer to proxy\n03b2b3b0 Add access to the resource attribute by server-side name\n2d16931c Update TOX_CONSTRAINTS_FILE for feature/r1\n7a19a5ff Update .gitreview for feature/r1\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Ie3ea1b21df8b20ca357e0457d39e9a411b27e784\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b05f1110bd8e999c055f354d5c5eee3053d2dcaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b05f1110bd8e999c055f354d5c5eee3053d2dcaf"}]},"branch":"refs/heads/master"},"d99d9cbe1dffd41cff1621a16382d015d657e22d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-05-16 10:20:04.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/44/841844/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/841844/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/841844/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/841844/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/841844/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/841844/2"}}},"commit":{"parents":[{"commit":"c9d01acbb6ee0b5ceaafd387932089df13a3f434","subject":"Merge \"[Puppet OpenStack] Transition stein to EOL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c9d01acbb6ee0b5ceaafd387932089df13a3f434"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-16 07:55:45.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2022-05-16 10:20:04.000000000","tz":0},"subject":"Release openstacksdk for Zed-1 milestone","message":"Release openstacksdk for Zed-1 milestone\n\nThis is a library release for openstacksdk for the Zed-1\nmilestone. This repository contains commits that have not been releases\nfor this cycle yet.\n\nIf the team is ready to process with this release, please leave a +1\nto indicate we should go ahead with the release.\n\nIf the team needs more time for things about to merge, or if there is\nsome other reason a release should not be done at this time, please\nleave a -1 with a comment indicating the status. Then update the patch\nwith the new commit hash to use once the team is ready to do the\nrelease.\n\n$ git log --oneline --no-merges 0.61.0..652faab5\n360d517a Improve StatsD metric precision\n1e50ea50 Add R1 summary release note\n6b2f555a image: Add \"id\" filter for images\n46db8d3b Allow to filter endpoints by region_id\n975cabbd Add QoS rule type filtering keys\nbdccc6b7 fix: improperly encoded object names\n358ddcd3 Run nodepool job\n53e81c16 Add Python3 zed unit tests\n194f3545 Update master for stable/yoga\nbdd87c70 Identity: Add support for system role assignment\n4c1ffee6 Indentation of the docstrings\n44c55156 network: Fix update of network provider\n55eff088 Correct documentation about Load Balancer API\nbcd5916c Improve compute flavor handling\ndd892d7b Use empty read_acl for swift container\n26a5d10d Restore get_compute_limits backward compatibility\n1b479f0b Revive legacy job\n7569cd9a Skip tests when needed extensions are disabled\nfe3d36f0 Cloud / Utils service - reindentation of the docstrings\n827736d3 Cloud / Network service - reindentation of the docstrings\n6650374b Cloud / Identity service - reindentation of the docstrings\n3244abd0 tests: Centralize configuration of default flavor, image\n2ee2df6b Skip qos-pps-minimum tests if extension is missing\n11d89450 doc: Update testing documentation\n05a83963 doc: Remove references to \u0027examples\u0027 test env\n1f1aa4fa Add query parameters to local ip\n2d69560e Cloud / Compute service - reindentation of the docstrings\n441b7d0c Cloud / Dns service - reindentation of the docstrings\n63aa7c00 Cloud / Coe service - reindentation of the docstrings\n0ce0fddc Cloud / Baremetal service - reindentation of the docstrings\n582e301b Cloud / Security group service - reindentation of the docstrings\n560e1f91 Cloud / Normalize service - reindentation of the docstrings\ndd9b95c1 Cloud / Floating service - reindentation of the docstrings\nafb49692 fix creation of protected image\n5e969fb4 Revert \"Add \"security_group_ids\" to Port\u0027s query parameters\"\n3c8826d0 Get rid of normalization in compute CL\ncadee6fa Functional tests for vpn ike policy resource\n3e84351c Add QoS min pps rule object and CRUD operations\n4f54021e Get rid of normalization in network CL\ned86ffbf Get rid of normalization in further CL services\ne910e6ea Get rid of normalization in identity CL\nf34ea0d1 Get rid of normalization in orchestration CL\n6c905301 Get rid of normalization for image service\nfc2d56fa Object Store - reindentation of the docstrings\nee24cbff Switch quota_set cloud layer BS methods to proxy\nefdc8c2a Switch unittests for BS to use v3\n82f09b4e Reindentation of the docstrings for baremetal service\n4831e22a Switch quota methods of cloud layer to proxy\n29c87246 Switch create_server cloud method to proxy\nc0a9b64e Switch delete_server cloud method to rely on proxy\n9d458860 tests: Handle overridden \u0027verify_delete\u0027 (kw)args\nf4dafb3a Switch update_server cloud layer to proxy\na92ee7ae Switch rebuild_server cloud method to rely on proxy\nd7098876 Fix misuse of assertTrue\nb2bea061 Identity service - reunite :class: links on single lines\n6775fb80 Block storage - reunite :class: links on single lines\ne94fda9e Switch cloud.compute.get_server_by_id to use proxy\n104a361a Add revert share to snapshot to shared file system.\n7c003e6a Shared file system - reunite :class: links on single lines\nda717351 Instance HA service - reunite :class: links on single lines\na1263076 Load balancer - reindentation of the docstrings\n831f7bd5 Network service - reunite :class: links on single lines\n9290eeea Orchestration service - reunite class links (:class:) on one line\n02c175db Compute - reindentation of the docstrings\nf5b60d2b Add Neutron Local IP CRUD\n94c6131c trivial: Correct docstrings for \u0027delete\u0027 proxy calls\n353a0ffe compute: Server group rules are dicts, not lists of dicts\n41893385 Fix functional test for user message\nb48d0b8c Database service - reindentation of the docstrings\n15da4d22 Image service - reindentation of the docstrings\naa58c884 Orchestration- reindentation of the docstrings\n42468112 Keep creating keystone admin endpoint for heat\nef1bee16 Switch compute limits cloud to proxy\n31c85038 Shared File System - reindentation of the docstrings\n72c223b5 Add share snapshot to shared file system.\nef587c9a compute: Add support for microversion 2.89\n770a3c3e compute: Add support for migrations API\n6d1321f5 compute: Add support for server migrations API\n6c96faa7 compute: Regroup proxy methods\nbf267b0e DNS - reindentation of the docstrings\ne574e4dd Clustering - reindentation of the docstrings\n0fc28b39 Network - reindentation of the docstrings\nb47652cf trivial: Typo fixes\n86c85103 Remove \u0027__unicode__\u0027 helper\n58270810 exception: Correct argument\n95640856 Add support for updated_at field for volume objects\n894b5450 Block storage - reindentation of the docstrings\n9d143ed8 Reindentation of the docstrings\n4581db6c Reindentation of the docstrings\nc746567a Reindentation of the docstrings\nef1b622b Reindentation of the docstrings\n322c72ba Reindentation of the docstrings\n4ad2d0d2 Reindentation of the docstrings\n388a463a Reindentation of the docstrings\n45f400dc resource: Rewrap function signatures\n93a8258e resource: Remove deprecated \u0027allow_get\u0027 attribute\n07eb09f6 Rely on proxy for compute.flavor cloud layer operations\n15a8fc4c Support description in sg-rule creation\n16e557f4 Switch Swift cloud layer to proxy\nfde91622 Splits class `TestSharedFileSystemProxy`\nd3b614ad Splits class `TestPlacementProxy`\n6bdde3d1 Splits class `TestOrchestrationProxy`\n93f69e42 Splits class `TestNetworkProxy`\n81776e61 Add missing headers in object store\n56068af2 Splits class `TestMessageProxy`\n45995c88 Splits class `TestKeyManagerProxy`\nf5f0f9f1 Splits class `TestInstanceHaProxy`\n50043daf Splits the `TestBaremetalProxy` class\n290a6cf5 Add limit resource to shared file system\nbba9c5c5 Fix setting initial object/container metadata.\n6a2420db Splits the `TestAcceleratorProxy` class\na2b7c540 Optimize code by moving misc functions to utils\n7f87d2ee Start splitting the `TestImageProxy` class\n4914d8d7 Add user message to shared file system\n67c8c191 Replace the \u0027try except\u0027 block\nd7d6464c Add user group assignment support in identity\n21795acf Add a manila functional check job\nd62a4e60 Fix key generation for caching\nf7f9446d Add \"security_group_ids\" to Port\u0027s query parameters\n88e520f5 Adds storage pools to shared file system\na8c3cb29 Extend has_version function to accept version parameter\n1909df1f Vpn ike policy resource\n3d6aff34 Add common metadata mixin to block storage and clustering\nc1c8dfb7 Make metadata a common mixin\nf3df2095 Introduce QuotaSet in block storage service\nb7ebc731 Introduce QuotaSet in the compute service\n94cb864d Introduce common class for QuotaSet\nfae7ab06 Fix import order in block_storage tests\n2d56b681 Fix import order in cloud tests\nc6494db0 Move tag mixin into the common\nb59cda5c Fix import order in clustering tests\nadcf98c2 Fix import order in orchestration tests\nb86d3a8c Fix import order in identity tests\n8a1d255c Switch identity roles in the cloud layer to proxy\n95cec287 Fix identity role management\n533ba93a Switch identiy.endpoints in the cloud layer\ndf5f987c Add share resource to shared file system\n9a038d8c Switch identity.service in cloud layer to proxy\ne192f215 Switch identity.domains in cloud layer\n54f910f2 Switch identity groups and users in cloud layer\n74fa9871 Add possibility to create subprojects in cloud layer\n72aa4531 Switch project management in cloud to proxy\nccc6478c Temporarily disable nodepool job\nd6c51c53 Add cleanup function for DNS resource\nf5c9653a Switch BS volume_attachment operations in cloud layer\n485b53bd Switch BS backup and snapshot methods in cloud layer to proxy\n734ae40f Add BS snapshot and backup actions\n118cf735 Switch BS type access operations in cloud layer\n8579b095 Add BS type access into proxy/resource layer\nd235b141 Switch block_storage.volume operations in cloud layer\nf9183cf4 Add block_storage.volume actions\nd1c9a96e Copy find_volume to block_storage.v2 proxy\na3582786 Fix active status for block storage\ne6fed6d2 Add scheduler hints on the block_storage.volume resource\n62781ae2 Sort block storage properties alphabetically\nc6177a47 Drop cloud layer methods for Senlin\n7cee92d5 Replace deprecated inspect.getargspec call\n28875355 Revert \"Update TOX_CONSTRAINTS_FILE for feature/r1\"\n698ac14a Switch object_store cloud functions to proxy layer\n70a06d99 Stop sending tenant_id to Neutron\n6db391f6 Switch networking function in cloud layer to proxy\n03b2b3b0 Add access to the resource attribute by server-side name\n2d16931c Update TOX_CONSTRAINTS_FILE for feature/r1\n7a19a5ff Update .gitreview for feature/r1\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Ie3ea1b21df8b20ca357e0457d39e9a411b27e784\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d99d9cbe1dffd41cff1621a16382d015d657e22d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d99d9cbe1dffd41cff1621a16382d015d657e22d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"561cf4dbf3a24a4d4507ff7b4052be3f36d318e8":{"kind":"REWORK","_number":3,"created":"2022-05-18 15:17:14.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/44/841844/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/841844/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/841844/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/841844/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/841844/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/841844/3"}}},"commit":{"parents":[{"commit":"c9d01acbb6ee0b5ceaafd387932089df13a3f434","subject":"Merge \"[Puppet OpenStack] Transition stein to EOL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c9d01acbb6ee0b5ceaafd387932089df13a3f434"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-16 07:55:45.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"Artem.goncharov@gmail.com","date":"2022-05-18 15:16:59.000000000","tz":120},"subject":"Release openstacksdk for Zed-1 milestone","message":"Release openstacksdk for Zed-1 milestone\n\nThis is a library release for openstacksdk for the Zed-1\nmilestone. This repository contains commits that have not been releases\nfor this cycle yet.\n\nIf the team is ready to process with this release, please leave a +1\nto indicate we should go ahead with the release.\n\nIf the team needs more time for things about to merge, or if there is\nsome other reason a release should not be done at this time, please\nleave a -1 with a comment indicating the status. Then update the patch\nwith the new commit hash to use once the team is ready to do the\nrelease.\n\n$ git log --oneline --no-merges 0.61.0..652faab5\n360d517a Improve StatsD metric precision\n1e50ea50 Add R1 summary release note\n6b2f555a image: Add \"id\" filter for images\n46db8d3b Allow to filter endpoints by region_id\n975cabbd Add QoS rule type filtering keys\nbdccc6b7 fix: improperly encoded object names\n358ddcd3 Run nodepool job\n53e81c16 Add Python3 zed unit tests\n194f3545 Update master for stable/yoga\nbdd87c70 Identity: Add support for system role assignment\n4c1ffee6 Indentation of the docstrings\n44c55156 network: Fix update of network provider\n55eff088 Correct documentation about Load Balancer API\nbcd5916c Improve compute flavor handling\ndd892d7b Use empty read_acl for swift container\n26a5d10d Restore get_compute_limits backward compatibility\n1b479f0b Revive legacy job\n7569cd9a Skip tests when needed extensions are disabled\nfe3d36f0 Cloud / Utils service - reindentation of the docstrings\n827736d3 Cloud / Network service - reindentation of the docstrings\n6650374b Cloud / Identity service - reindentation of the docstrings\n3244abd0 tests: Centralize configuration of default flavor, image\n2ee2df6b Skip qos-pps-minimum tests if extension is missing\n11d89450 doc: Update testing documentation\n05a83963 doc: Remove references to \u0027examples\u0027 test env\n1f1aa4fa Add query parameters to local ip\n2d69560e Cloud / Compute service - reindentation of the docstrings\n441b7d0c Cloud / Dns service - reindentation of the docstrings\n63aa7c00 Cloud / Coe service - reindentation of the docstrings\n0ce0fddc Cloud / Baremetal service - reindentation of the docstrings\n582e301b Cloud / Security group service - reindentation of the docstrings\n560e1f91 Cloud / Normalize service - reindentation of the docstrings\ndd9b95c1 Cloud / Floating service - reindentation of the docstrings\nafb49692 fix creation of protected image\n5e969fb4 Revert \"Add \"security_group_ids\" to Port\u0027s query parameters\"\n3c8826d0 Get rid of normalization in compute CL\ncadee6fa Functional tests for vpn ike policy resource\n3e84351c Add QoS min pps rule object and CRUD operations\n4f54021e Get rid of normalization in network CL\ned86ffbf Get rid of normalization in further CL services\ne910e6ea Get rid of normalization in identity CL\nf34ea0d1 Get rid of normalization in orchestration CL\n6c905301 Get rid of normalization for image service\nfc2d56fa Object Store - reindentation of the docstrings\nee24cbff Switch quota_set cloud layer BS methods to proxy\nefdc8c2a Switch unittests for BS to use v3\n82f09b4e Reindentation of the docstrings for baremetal service\n4831e22a Switch quota methods of cloud layer to proxy\n29c87246 Switch create_server cloud method to proxy\nc0a9b64e Switch delete_server cloud method to rely on proxy\n9d458860 tests: Handle overridden \u0027verify_delete\u0027 (kw)args\nf4dafb3a Switch update_server cloud layer to proxy\na92ee7ae Switch rebuild_server cloud method to rely on proxy\nd7098876 Fix misuse of assertTrue\nb2bea061 Identity service - reunite :class: links on single lines\n6775fb80 Block storage - reunite :class: links on single lines\ne94fda9e Switch cloud.compute.get_server_by_id to use proxy\n104a361a Add revert share to snapshot to shared file system.\n7c003e6a Shared file system - reunite :class: links on single lines\nda717351 Instance HA service - reunite :class: links on single lines\na1263076 Load balancer - reindentation of the docstrings\n831f7bd5 Network service - reunite :class: links on single lines\n9290eeea Orchestration service - reunite class links (:class:) on one line\n02c175db Compute - reindentation of the docstrings\nf5b60d2b Add Neutron Local IP CRUD\n94c6131c trivial: Correct docstrings for \u0027delete\u0027 proxy calls\n353a0ffe compute: Server group rules are dicts, not lists of dicts\n41893385 Fix functional test for user message\nb48d0b8c Database service - reindentation of the docstrings\n15da4d22 Image service - reindentation of the docstrings\naa58c884 Orchestration- reindentation of the docstrings\n42468112 Keep creating keystone admin endpoint for heat\nef1bee16 Switch compute limits cloud to proxy\n31c85038 Shared File System - reindentation of the docstrings\n72c223b5 Add share snapshot to shared file system.\nef587c9a compute: Add support for microversion 2.89\n770a3c3e compute: Add support for migrations API\n6d1321f5 compute: Add support for server migrations API\n6c96faa7 compute: Regroup proxy methods\nbf267b0e DNS - reindentation of the docstrings\ne574e4dd Clustering - reindentation of the docstrings\n0fc28b39 Network - reindentation of the docstrings\nb47652cf trivial: Typo fixes\n86c85103 Remove \u0027__unicode__\u0027 helper\n58270810 exception: Correct argument\n95640856 Add support for updated_at field for volume objects\n894b5450 Block storage - reindentation of the docstrings\n9d143ed8 Reindentation of the docstrings\n4581db6c Reindentation of the docstrings\nc746567a Reindentation of the docstrings\nef1b622b Reindentation of the docstrings\n322c72ba Reindentation of the docstrings\n4ad2d0d2 Reindentation of the docstrings\n388a463a Reindentation of the docstrings\n45f400dc resource: Rewrap function signatures\n93a8258e resource: Remove deprecated \u0027allow_get\u0027 attribute\n07eb09f6 Rely on proxy for compute.flavor cloud layer operations\n15a8fc4c Support description in sg-rule creation\n16e557f4 Switch Swift cloud layer to proxy\nfde91622 Splits class `TestSharedFileSystemProxy`\nd3b614ad Splits class `TestPlacementProxy`\n6bdde3d1 Splits class `TestOrchestrationProxy`\n93f69e42 Splits class `TestNetworkProxy`\n81776e61 Add missing headers in object store\n56068af2 Splits class `TestMessageProxy`\n45995c88 Splits class `TestKeyManagerProxy`\nf5f0f9f1 Splits class `TestInstanceHaProxy`\n50043daf Splits the `TestBaremetalProxy` class\n290a6cf5 Add limit resource to shared file system\nbba9c5c5 Fix setting initial object/container metadata.\n6a2420db Splits the `TestAcceleratorProxy` class\na2b7c540 Optimize code by moving misc functions to utils\n7f87d2ee Start splitting the `TestImageProxy` class\n4914d8d7 Add user message to shared file system\n67c8c191 Replace the \u0027try except\u0027 block\nd7d6464c Add user group assignment support in identity\n21795acf Add a manila functional check job\nd62a4e60 Fix key generation for caching\nf7f9446d Add \"security_group_ids\" to Port\u0027s query parameters\n88e520f5 Adds storage pools to shared file system\na8c3cb29 Extend has_version function to accept version parameter\n1909df1f Vpn ike policy resource\n3d6aff34 Add common metadata mixin to block storage and clustering\nc1c8dfb7 Make metadata a common mixin\nf3df2095 Introduce QuotaSet in block storage service\nb7ebc731 Introduce QuotaSet in the compute service\n94cb864d Introduce common class for QuotaSet\nfae7ab06 Fix import order in block_storage tests\n2d56b681 Fix import order in cloud tests\nc6494db0 Move tag mixin into the common\nb59cda5c Fix import order in clustering tests\nadcf98c2 Fix import order in orchestration tests\nb86d3a8c Fix import order in identity tests\n8a1d255c Switch identity roles in the cloud layer to proxy\n95cec287 Fix identity role management\n533ba93a Switch identiy.endpoints in the cloud layer\ndf5f987c Add share resource to shared file system\n9a038d8c Switch identity.service in cloud layer to proxy\ne192f215 Switch identity.domains in cloud layer\n54f910f2 Switch identity groups and users in cloud layer\n74fa9871 Add possibility to create subprojects in cloud layer\n72aa4531 Switch project management in cloud to proxy\nccc6478c Temporarily disable nodepool job\nd6c51c53 Add cleanup function for DNS resource\nf5c9653a Switch BS volume_attachment operations in cloud layer\n485b53bd Switch BS backup and snapshot methods in cloud layer to proxy\n734ae40f Add BS snapshot and backup actions\n118cf735 Switch BS type access operations in cloud layer\n8579b095 Add BS type access into proxy/resource layer\nd235b141 Switch block_storage.volume operations in cloud layer\nf9183cf4 Add block_storage.volume actions\nd1c9a96e Copy find_volume to block_storage.v2 proxy\na3582786 Fix active status for block storage\ne6fed6d2 Add scheduler hints on the block_storage.volume resource\n62781ae2 Sort block storage properties alphabetically\nc6177a47 Drop cloud layer methods for Senlin\n7cee92d5 Replace deprecated inspect.getargspec call\n28875355 Revert \"Update TOX_CONSTRAINTS_FILE for feature/r1\"\n698ac14a Switch object_store cloud functions to proxy layer\n70a06d99 Stop sending tenant_id to Neutron\n6db391f6 Switch networking function in cloud layer to proxy\n03b2b3b0 Add access to the resource attribute by server-side name\n2d16931c Update TOX_CONSTRAINTS_FILE for feature/r1\n7a19a5ff Update .gitreview for feature/r1\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Ie3ea1b21df8b20ca357e0457d39e9a411b27e784\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/561cf4dbf3a24a4d4507ff7b4052be3f36d318e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/561cf4dbf3a24a4d4507ff7b4052be3f36d318e8"}]},"branch":"refs/heads/master"},"fc7f5538edae0e0d14eb27e98b15c4017c37b9e0":{"kind":"REWORK","_number":4,"created":"2022-05-18 16:33:42.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/44/841844/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/841844/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/841844/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/841844/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/841844/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/841844/4"}}},"commit":{"parents":[{"commit":"c9d01acbb6ee0b5ceaafd387932089df13a3f434","subject":"Merge \"[Puppet OpenStack] Transition stein to EOL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c9d01acbb6ee0b5ceaafd387932089df13a3f434"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-16 07:55:45.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"Artem.goncharov@gmail.com","date":"2022-05-18 16:33:00.000000000","tz":120},"subject":"Release openstacksdk for Zed-1 milestone","message":"Release openstacksdk for Zed-1 milestone\n\nThis is a library release for openstacksdk for the Zed-1\nmilestone. This repository contains commits that have not been releases\nfor this cycle yet.\n\nWith this OpenStackSDK is also changing the release model to be\nindependent.\n\nIf the team is ready to process with this release, please leave a +1\nto indicate we should go ahead with the release.\n\nIf the team needs more time for things about to merge, or if there is\nsome other reason a release should not be done at this time, please\nleave a -1 with a comment indicating the status. Then update the patch\nwith the new commit hash to use once the team is ready to do the\nrelease.\n\n$ git log --oneline --no-merges 0.61.0..652faab5\n360d517a Improve StatsD metric precision\n1e50ea50 Add R1 summary release note\n6b2f555a image: Add \"id\" filter for images\n46db8d3b Allow to filter endpoints by region_id\n975cabbd Add QoS rule type filtering keys\nbdccc6b7 fix: improperly encoded object names\n358ddcd3 Run nodepool job\n53e81c16 Add Python3 zed unit tests\n194f3545 Update master for stable/yoga\nbdd87c70 Identity: Add support for system role assignment\n4c1ffee6 Indentation of the docstrings\n44c55156 network: Fix update of network provider\n55eff088 Correct documentation about Load Balancer API\nbcd5916c Improve compute flavor handling\ndd892d7b Use empty read_acl for swift container\n26a5d10d Restore get_compute_limits backward compatibility\n1b479f0b Revive legacy job\n7569cd9a Skip tests when needed extensions are disabled\nfe3d36f0 Cloud / Utils service - reindentation of the docstrings\n827736d3 Cloud / Network service - reindentation of the docstrings\n6650374b Cloud / Identity service - reindentation of the docstrings\n3244abd0 tests: Centralize configuration of default flavor, image\n2ee2df6b Skip qos-pps-minimum tests if extension is missing\n11d89450 doc: Update testing documentation\n05a83963 doc: Remove references to \u0027examples\u0027 test env\n1f1aa4fa Add query parameters to local ip\n2d69560e Cloud / Compute service - reindentation of the docstrings\n441b7d0c Cloud / Dns service - reindentation of the docstrings\n63aa7c00 Cloud / Coe service - reindentation of the docstrings\n0ce0fddc Cloud / Baremetal service - reindentation of the docstrings\n582e301b Cloud / Security group service - reindentation of the docstrings\n560e1f91 Cloud / Normalize service - reindentation of the docstrings\ndd9b95c1 Cloud / Floating service - reindentation of the docstrings\nafb49692 fix creation of protected image\n5e969fb4 Revert \"Add \"security_group_ids\" to Port\u0027s query parameters\"\n3c8826d0 Get rid of normalization in compute CL\ncadee6fa Functional tests for vpn ike policy resource\n3e84351c Add QoS min pps rule object and CRUD operations\n4f54021e Get rid of normalization in network CL\ned86ffbf Get rid of normalization in further CL services\ne910e6ea Get rid of normalization in identity CL\nf34ea0d1 Get rid of normalization in orchestration CL\n6c905301 Get rid of normalization for image service\nfc2d56fa Object Store - reindentation of the docstrings\nee24cbff Switch quota_set cloud layer BS methods to proxy\nefdc8c2a Switch unittests for BS to use v3\n82f09b4e Reindentation of the docstrings for baremetal service\n4831e22a Switch quota methods of cloud layer to proxy\n29c87246 Switch create_server cloud method to proxy\nc0a9b64e Switch delete_server cloud method to rely on proxy\n9d458860 tests: Handle overridden \u0027verify_delete\u0027 (kw)args\nf4dafb3a Switch update_server cloud layer to proxy\na92ee7ae Switch rebuild_server cloud method to rely on proxy\nd7098876 Fix misuse of assertTrue\nb2bea061 Identity service - reunite :class: links on single lines\n6775fb80 Block storage - reunite :class: links on single lines\ne94fda9e Switch cloud.compute.get_server_by_id to use proxy\n104a361a Add revert share to snapshot to shared file system.\n7c003e6a Shared file system - reunite :class: links on single lines\nda717351 Instance HA service - reunite :class: links on single lines\na1263076 Load balancer - reindentation of the docstrings\n831f7bd5 Network service - reunite :class: links on single lines\n9290eeea Orchestration service - reunite class links (:class:) on one line\n02c175db Compute - reindentation of the docstrings\nf5b60d2b Add Neutron Local IP CRUD\n94c6131c trivial: Correct docstrings for \u0027delete\u0027 proxy calls\n353a0ffe compute: Server group rules are dicts, not lists of dicts\n41893385 Fix functional test for user message\nb48d0b8c Database service - reindentation of the docstrings\n15da4d22 Image service - reindentation of the docstrings\naa58c884 Orchestration- reindentation of the docstrings\n42468112 Keep creating keystone admin endpoint for heat\nef1bee16 Switch compute limits cloud to proxy\n31c85038 Shared File System - reindentation of the docstrings\n72c223b5 Add share snapshot to shared file system.\nef587c9a compute: Add support for microversion 2.89\n770a3c3e compute: Add support for migrations API\n6d1321f5 compute: Add support for server migrations API\n6c96faa7 compute: Regroup proxy methods\nbf267b0e DNS - reindentation of the docstrings\ne574e4dd Clustering - reindentation of the docstrings\n0fc28b39 Network - reindentation of the docstrings\nb47652cf trivial: Typo fixes\n86c85103 Remove \u0027__unicode__\u0027 helper\n58270810 exception: Correct argument\n95640856 Add support for updated_at field for volume objects\n894b5450 Block storage - reindentation of the docstrings\n9d143ed8 Reindentation of the docstrings\n4581db6c Reindentation of the docstrings\nc746567a Reindentation of the docstrings\nef1b622b Reindentation of the docstrings\n322c72ba Reindentation of the docstrings\n4ad2d0d2 Reindentation of the docstrings\n388a463a Reindentation of the docstrings\n45f400dc resource: Rewrap function signatures\n93a8258e resource: Remove deprecated \u0027allow_get\u0027 attribute\n07eb09f6 Rely on proxy for compute.flavor cloud layer operations\n15a8fc4c Support description in sg-rule creation\n16e557f4 Switch Swift cloud layer to proxy\nfde91622 Splits class `TestSharedFileSystemProxy`\nd3b614ad Splits class `TestPlacementProxy`\n6bdde3d1 Splits class `TestOrchestrationProxy`\n93f69e42 Splits class `TestNetworkProxy`\n81776e61 Add missing headers in object store\n56068af2 Splits class `TestMessageProxy`\n45995c88 Splits class `TestKeyManagerProxy`\nf5f0f9f1 Splits class `TestInstanceHaProxy`\n50043daf Splits the `TestBaremetalProxy` class\n290a6cf5 Add limit resource to shared file system\nbba9c5c5 Fix setting initial object/container metadata.\n6a2420db Splits the `TestAcceleratorProxy` class\na2b7c540 Optimize code by moving misc functions to utils\n7f87d2ee Start splitting the `TestImageProxy` class\n4914d8d7 Add user message to shared file system\n67c8c191 Replace the \u0027try except\u0027 block\nd7d6464c Add user group assignment support in identity\n21795acf Add a manila functional check job\nd62a4e60 Fix key generation for caching\nf7f9446d Add \"security_group_ids\" to Port\u0027s query parameters\n88e520f5 Adds storage pools to shared file system\na8c3cb29 Extend has_version function to accept version parameter\n1909df1f Vpn ike policy resource\n3d6aff34 Add common metadata mixin to block storage and clustering\nc1c8dfb7 Make metadata a common mixin\nf3df2095 Introduce QuotaSet in block storage service\nb7ebc731 Introduce QuotaSet in the compute service\n94cb864d Introduce common class for QuotaSet\nfae7ab06 Fix import order in block_storage tests\n2d56b681 Fix import order in cloud tests\nc6494db0 Move tag mixin into the common\nb59cda5c Fix import order in clustering tests\nadcf98c2 Fix import order in orchestration tests\nb86d3a8c Fix import order in identity tests\n8a1d255c Switch identity roles in the cloud layer to proxy\n95cec287 Fix identity role management\n533ba93a Switch identiy.endpoints in the cloud layer\ndf5f987c Add share resource to shared file system\n9a038d8c Switch identity.service in cloud layer to proxy\ne192f215 Switch identity.domains in cloud layer\n54f910f2 Switch identity groups and users in cloud layer\n74fa9871 Add possibility to create subprojects in cloud layer\n72aa4531 Switch project management in cloud to proxy\nccc6478c Temporarily disable nodepool job\nd6c51c53 Add cleanup function for DNS resource\nf5c9653a Switch BS volume_attachment operations in cloud layer\n485b53bd Switch BS backup and snapshot methods in cloud layer to proxy\n734ae40f Add BS snapshot and backup actions\n118cf735 Switch BS type access operations in cloud layer\n8579b095 Add BS type access into proxy/resource layer\nd235b141 Switch block_storage.volume operations in cloud layer\nf9183cf4 Add block_storage.volume actions\nd1c9a96e Copy find_volume to block_storage.v2 proxy\na3582786 Fix active status for block storage\ne6fed6d2 Add scheduler hints on the block_storage.volume resource\n62781ae2 Sort block storage properties alphabetically\nc6177a47 Drop cloud layer methods for Senlin\n7cee92d5 Replace deprecated inspect.getargspec call\n28875355 Revert \"Update TOX_CONSTRAINTS_FILE for feature/r1\"\n698ac14a Switch object_store cloud functions to proxy layer\n70a06d99 Stop sending tenant_id to Neutron\n6db391f6 Switch networking function in cloud layer to proxy\n03b2b3b0 Add access to the resource attribute by server-side name\n2d16931c Update TOX_CONSTRAINTS_FILE for feature/r1\n7a19a5ff Update .gitreview for feature/r1\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Ie3ea1b21df8b20ca357e0457d39e9a411b27e784\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fc7f5538edae0e0d14eb27e98b15c4017c37b9e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fc7f5538edae0e0d14eb27e98b15c4017c37b9e0"}]},"branch":"refs/heads/master"},"20e6409c81e8dd3925cfd4562df5225540bfc3fb":{"kind":"REWORK","_number":5,"created":"2022-05-18 18:32:00.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/44/841844/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/841844/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/841844/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/841844/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/841844/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/841844/5"}}},"commit":{"parents":[{"commit":"c9d01acbb6ee0b5ceaafd387932089df13a3f434","subject":"Merge \"[Puppet OpenStack] Transition stein to EOL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c9d01acbb6ee0b5ceaafd387932089df13a3f434"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-16 07:55:45.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-05-18 18:31:09.000000000","tz":120},"subject":"Release openstacksdk for Zed-1 milestone","message":"Release openstacksdk for Zed-1 milestone\n\nThis is a library release for openstacksdk for the Zed-1\nmilestone. This repository contains commits that have not been released\nfor this cycle yet.\n\nWith this OpenStackSDK is also changing the release model to be\ncycle-with-rc.\n\nIf the team is ready to process with this release, please leave a +1\nto indicate we should go ahead with the release.\n\nIf the team needs more time for things about to merge, or if there is\nsome other reason a release should not be done at this time, please\nleave a -1 with a comment indicating the status. Then update the patch\nwith the new commit hash to use once the team is ready to do the\nrelease.\n\n$ git log --oneline --no-merges 0.61.0..652faab5\n360d517a Improve StatsD metric precision\n1e50ea50 Add R1 summary release note\n6b2f555a image: Add \"id\" filter for images\n46db8d3b Allow to filter endpoints by region_id\n975cabbd Add QoS rule type filtering keys\nbdccc6b7 fix: improperly encoded object names\n358ddcd3 Run nodepool job\n53e81c16 Add Python3 zed unit tests\n194f3545 Update master for stable/yoga\nbdd87c70 Identity: Add support for system role assignment\n4c1ffee6 Indentation of the docstrings\n44c55156 network: Fix update of network provider\n55eff088 Correct documentation about Load Balancer API\nbcd5916c Improve compute flavor handling\ndd892d7b Use empty read_acl for swift container\n26a5d10d Restore get_compute_limits backward compatibility\n1b479f0b Revive legacy job\n7569cd9a Skip tests when needed extensions are disabled\nfe3d36f0 Cloud / Utils service - reindentation of the docstrings\n827736d3 Cloud / Network service - reindentation of the docstrings\n6650374b Cloud / Identity service - reindentation of the docstrings\n3244abd0 tests: Centralize configuration of default flavor, image\n2ee2df6b Skip qos-pps-minimum tests if extension is missing\n11d89450 doc: Update testing documentation\n05a83963 doc: Remove references to \u0027examples\u0027 test env\n1f1aa4fa Add query parameters to local ip\n2d69560e Cloud / Compute service - reindentation of the docstrings\n441b7d0c Cloud / Dns service - reindentation of the docstrings\n63aa7c00 Cloud / Coe service - reindentation of the docstrings\n0ce0fddc Cloud / Baremetal service - reindentation of the docstrings\n582e301b Cloud / Security group service - reindentation of the docstrings\n560e1f91 Cloud / Normalize service - reindentation of the docstrings\ndd9b95c1 Cloud / Floating service - reindentation of the docstrings\nafb49692 fix creation of protected image\n5e969fb4 Revert \"Add \"security_group_ids\" to Port\u0027s query parameters\"\n3c8826d0 Get rid of normalization in compute CL\ncadee6fa Functional tests for vpn ike policy resource\n3e84351c Add QoS min pps rule object and CRUD operations\n4f54021e Get rid of normalization in network CL\ned86ffbf Get rid of normalization in further CL services\ne910e6ea Get rid of normalization in identity CL\nf34ea0d1 Get rid of normalization in orchestration CL\n6c905301 Get rid of normalization for image service\nfc2d56fa Object Store - reindentation of the docstrings\nee24cbff Switch quota_set cloud layer BS methods to proxy\nefdc8c2a Switch unittests for BS to use v3\n82f09b4e Reindentation of the docstrings for baremetal service\n4831e22a Switch quota methods of cloud layer to proxy\n29c87246 Switch create_server cloud method to proxy\nc0a9b64e Switch delete_server cloud method to rely on proxy\n9d458860 tests: Handle overridden \u0027verify_delete\u0027 (kw)args\nf4dafb3a Switch update_server cloud layer to proxy\na92ee7ae Switch rebuild_server cloud method to rely on proxy\nd7098876 Fix misuse of assertTrue\nb2bea061 Identity service - reunite :class: links on single lines\n6775fb80 Block storage - reunite :class: links on single lines\ne94fda9e Switch cloud.compute.get_server_by_id to use proxy\n104a361a Add revert share to snapshot to shared file system.\n7c003e6a Shared file system - reunite :class: links on single lines\nda717351 Instance HA service - reunite :class: links on single lines\na1263076 Load balancer - reindentation of the docstrings\n831f7bd5 Network service - reunite :class: links on single lines\n9290eeea Orchestration service - reunite class links (:class:) on one line\n02c175db Compute - reindentation of the docstrings\nf5b60d2b Add Neutron Local IP CRUD\n94c6131c trivial: Correct docstrings for \u0027delete\u0027 proxy calls\n353a0ffe compute: Server group rules are dicts, not lists of dicts\n41893385 Fix functional test for user message\nb48d0b8c Database service - reindentation of the docstrings\n15da4d22 Image service - reindentation of the docstrings\naa58c884 Orchestration- reindentation of the docstrings\n42468112 Keep creating keystone admin endpoint for heat\nef1bee16 Switch compute limits cloud to proxy\n31c85038 Shared File System - reindentation of the docstrings\n72c223b5 Add share snapshot to shared file system.\nef587c9a compute: Add support for microversion 2.89\n770a3c3e compute: Add support for migrations API\n6d1321f5 compute: Add support for server migrations API\n6c96faa7 compute: Regroup proxy methods\nbf267b0e DNS - reindentation of the docstrings\ne574e4dd Clustering - reindentation of the docstrings\n0fc28b39 Network - reindentation of the docstrings\nb47652cf trivial: Typo fixes\n86c85103 Remove \u0027__unicode__\u0027 helper\n58270810 exception: Correct argument\n95640856 Add support for updated_at field for volume objects\n894b5450 Block storage - reindentation of the docstrings\n9d143ed8 Reindentation of the docstrings\n4581db6c Reindentation of the docstrings\nc746567a Reindentation of the docstrings\nef1b622b Reindentation of the docstrings\n322c72ba Reindentation of the docstrings\n4ad2d0d2 Reindentation of the docstrings\n388a463a Reindentation of the docstrings\n45f400dc resource: Rewrap function signatures\n93a8258e resource: Remove deprecated \u0027allow_get\u0027 attribute\n07eb09f6 Rely on proxy for compute.flavor cloud layer operations\n15a8fc4c Support description in sg-rule creation\n16e557f4 Switch Swift cloud layer to proxy\nfde91622 Splits class `TestSharedFileSystemProxy`\nd3b614ad Splits class `TestPlacementProxy`\n6bdde3d1 Splits class `TestOrchestrationProxy`\n93f69e42 Splits class `TestNetworkProxy`\n81776e61 Add missing headers in object store\n56068af2 Splits class `TestMessageProxy`\n45995c88 Splits class `TestKeyManagerProxy`\nf5f0f9f1 Splits class `TestInstanceHaProxy`\n50043daf Splits the `TestBaremetalProxy` class\n290a6cf5 Add limit resource to shared file system\nbba9c5c5 Fix setting initial object/container metadata.\n6a2420db Splits the `TestAcceleratorProxy` class\na2b7c540 Optimize code by moving misc functions to utils\n7f87d2ee Start splitting the `TestImageProxy` class\n4914d8d7 Add user message to shared file system\n67c8c191 Replace the \u0027try except\u0027 block\nd7d6464c Add user group assignment support in identity\n21795acf Add a manila functional check job\nd62a4e60 Fix key generation for caching\nf7f9446d Add \"security_group_ids\" to Port\u0027s query parameters\n88e520f5 Adds storage pools to shared file system\na8c3cb29 Extend has_version function to accept version parameter\n1909df1f Vpn ike policy resource\n3d6aff34 Add common metadata mixin to block storage and clustering\nc1c8dfb7 Make metadata a common mixin\nf3df2095 Introduce QuotaSet in block storage service\nb7ebc731 Introduce QuotaSet in the compute service\n94cb864d Introduce common class for QuotaSet\nfae7ab06 Fix import order in block_storage tests\n2d56b681 Fix import order in cloud tests\nc6494db0 Move tag mixin into the common\nb59cda5c Fix import order in clustering tests\nadcf98c2 Fix import order in orchestration tests\nb86d3a8c Fix import order in identity tests\n8a1d255c Switch identity roles in the cloud layer to proxy\n95cec287 Fix identity role management\n533ba93a Switch identiy.endpoints in the cloud layer\ndf5f987c Add share resource to shared file system\n9a038d8c Switch identity.service in cloud layer to proxy\ne192f215 Switch identity.domains in cloud layer\n54f910f2 Switch identity groups and users in cloud layer\n74fa9871 Add possibility to create subprojects in cloud layer\n72aa4531 Switch project management in cloud to proxy\nccc6478c Temporarily disable nodepool job\nd6c51c53 Add cleanup function for DNS resource\nf5c9653a Switch BS volume_attachment operations in cloud layer\n485b53bd Switch BS backup and snapshot methods in cloud layer to proxy\n734ae40f Add BS snapshot and backup actions\n118cf735 Switch BS type access operations in cloud layer\n8579b095 Add BS type access into proxy/resource layer\nd235b141 Switch block_storage.volume operations in cloud layer\nf9183cf4 Add block_storage.volume actions\nd1c9a96e Copy find_volume to block_storage.v2 proxy\na3582786 Fix active status for block storage\ne6fed6d2 Add scheduler hints on the block_storage.volume resource\n62781ae2 Sort block storage properties alphabetically\nc6177a47 Drop cloud layer methods for Senlin\n7cee92d5 Replace deprecated inspect.getargspec call\n28875355 Revert \"Update TOX_CONSTRAINTS_FILE for feature/r1\"\n698ac14a Switch object_store cloud functions to proxy layer\n70a06d99 Stop sending tenant_id to Neutron\n6db391f6 Switch networking function in cloud layer to proxy\n03b2b3b0 Add access to the resource attribute by server-side name\n2d16931c Update TOX_CONSTRAINTS_FILE for feature/r1\n7a19a5ff Update .gitreview for feature/r1\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Ie3ea1b21df8b20ca357e0457d39e9a411b27e784\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/20e6409c81e8dd3925cfd4562df5225540bfc3fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/20e6409c81e8dd3925cfd4562df5225540bfc3fb"}]},"branch":"refs/heads/master"},"7934e97d79437c4297645cdc49909205fd42397a":{"kind":"REWORK","_number":6,"created":"2022-05-19 09:53:05.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/44/841844/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/841844/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/841844/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/841844/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/841844/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/841844/6"}}},"commit":{"parents":[{"commit":"c9d01acbb6ee0b5ceaafd387932089df13a3f434","subject":"Merge \"[Puppet OpenStack] Transition stein to EOL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c9d01acbb6ee0b5ceaafd387932089df13a3f434"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-16 07:55:45.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-05-19 09:49:42.000000000","tz":120},"subject":"Release openstacksdk for Zed-1 milestone","message":"Release openstacksdk for Zed-1 milestone\n\nThis is a library release for openstacksdk for the Zed-1\nmilestone. This repository contains commits that have not been released\nfor this cycle yet.\n\nIf the team is ready to process with this release, please leave a +1\nto indicate we should go ahead with the release.\n\nIf the team needs more time for things about to merge, or if there is\nsome other reason a release should not be done at this time, please\nleave a -1 with a comment indicating the status. Then update the patch\nwith the new commit hash to use once the team is ready to do the\nrelease.\n\n$ git log --oneline --no-merges 0.61.0..652faab5\n360d517a Improve StatsD metric precision\n1e50ea50 Add R1 summary release note\n6b2f555a image: Add \"id\" filter for images\n46db8d3b Allow to filter endpoints by region_id\n975cabbd Add QoS rule type filtering keys\nbdccc6b7 fix: improperly encoded object names\n358ddcd3 Run nodepool job\n53e81c16 Add Python3 zed unit tests\n194f3545 Update master for stable/yoga\nbdd87c70 Identity: Add support for system role assignment\n4c1ffee6 Indentation of the docstrings\n44c55156 network: Fix update of network provider\n55eff088 Correct documentation about Load Balancer API\nbcd5916c Improve compute flavor handling\ndd892d7b Use empty read_acl for swift container\n26a5d10d Restore get_compute_limits backward compatibility\n1b479f0b Revive legacy job\n7569cd9a Skip tests when needed extensions are disabled\nfe3d36f0 Cloud / Utils service - reindentation of the docstrings\n827736d3 Cloud / Network service - reindentation of the docstrings\n6650374b Cloud / Identity service - reindentation of the docstrings\n3244abd0 tests: Centralize configuration of default flavor, image\n2ee2df6b Skip qos-pps-minimum tests if extension is missing\n11d89450 doc: Update testing documentation\n05a83963 doc: Remove references to \u0027examples\u0027 test env\n1f1aa4fa Add query parameters to local ip\n2d69560e Cloud / Compute service - reindentation of the docstrings\n441b7d0c Cloud / Dns service - reindentation of the docstrings\n63aa7c00 Cloud / Coe service - reindentation of the docstrings\n0ce0fddc Cloud / Baremetal service - reindentation of the docstrings\n582e301b Cloud / Security group service - reindentation of the docstrings\n560e1f91 Cloud / Normalize service - reindentation of the docstrings\ndd9b95c1 Cloud / Floating service - reindentation of the docstrings\nafb49692 fix creation of protected image\n5e969fb4 Revert \"Add \"security_group_ids\" to Port\u0027s query parameters\"\n3c8826d0 Get rid of normalization in compute CL\ncadee6fa Functional tests for vpn ike policy resource\n3e84351c Add QoS min pps rule object and CRUD operations\n4f54021e Get rid of normalization in network CL\ned86ffbf Get rid of normalization in further CL services\ne910e6ea Get rid of normalization in identity CL\nf34ea0d1 Get rid of normalization in orchestration CL\n6c905301 Get rid of normalization for image service\nfc2d56fa Object Store - reindentation of the docstrings\nee24cbff Switch quota_set cloud layer BS methods to proxy\nefdc8c2a Switch unittests for BS to use v3\n82f09b4e Reindentation of the docstrings for baremetal service\n4831e22a Switch quota methods of cloud layer to proxy\n29c87246 Switch create_server cloud method to proxy\nc0a9b64e Switch delete_server cloud method to rely on proxy\n9d458860 tests: Handle overridden \u0027verify_delete\u0027 (kw)args\nf4dafb3a Switch update_server cloud layer to proxy\na92ee7ae Switch rebuild_server cloud method to rely on proxy\nd7098876 Fix misuse of assertTrue\nb2bea061 Identity service - reunite :class: links on single lines\n6775fb80 Block storage - reunite :class: links on single lines\ne94fda9e Switch cloud.compute.get_server_by_id to use proxy\n104a361a Add revert share to snapshot to shared file system.\n7c003e6a Shared file system - reunite :class: links on single lines\nda717351 Instance HA service - reunite :class: links on single lines\na1263076 Load balancer - reindentation of the docstrings\n831f7bd5 Network service - reunite :class: links on single lines\n9290eeea Orchestration service - reunite class links (:class:) on one line\n02c175db Compute - reindentation of the docstrings\nf5b60d2b Add Neutron Local IP CRUD\n94c6131c trivial: Correct docstrings for \u0027delete\u0027 proxy calls\n353a0ffe compute: Server group rules are dicts, not lists of dicts\n41893385 Fix functional test for user message\nb48d0b8c Database service - reindentation of the docstrings\n15da4d22 Image service - reindentation of the docstrings\naa58c884 Orchestration- reindentation of the docstrings\n42468112 Keep creating keystone admin endpoint for heat\nef1bee16 Switch compute limits cloud to proxy\n31c85038 Shared File System - reindentation of the docstrings\n72c223b5 Add share snapshot to shared file system.\nef587c9a compute: Add support for microversion 2.89\n770a3c3e compute: Add support for migrations API\n6d1321f5 compute: Add support for server migrations API\n6c96faa7 compute: Regroup proxy methods\nbf267b0e DNS - reindentation of the docstrings\ne574e4dd Clustering - reindentation of the docstrings\n0fc28b39 Network - reindentation of the docstrings\nb47652cf trivial: Typo fixes\n86c85103 Remove \u0027__unicode__\u0027 helper\n58270810 exception: Correct argument\n95640856 Add support for updated_at field for volume objects\n894b5450 Block storage - reindentation of the docstrings\n9d143ed8 Reindentation of the docstrings\n4581db6c Reindentation of the docstrings\nc746567a Reindentation of the docstrings\nef1b622b Reindentation of the docstrings\n322c72ba Reindentation of the docstrings\n4ad2d0d2 Reindentation of the docstrings\n388a463a Reindentation of the docstrings\n45f400dc resource: Rewrap function signatures\n93a8258e resource: Remove deprecated \u0027allow_get\u0027 attribute\n07eb09f6 Rely on proxy for compute.flavor cloud layer operations\n15a8fc4c Support description in sg-rule creation\n16e557f4 Switch Swift cloud layer to proxy\nfde91622 Splits class `TestSharedFileSystemProxy`\nd3b614ad Splits class `TestPlacementProxy`\n6bdde3d1 Splits class `TestOrchestrationProxy`\n93f69e42 Splits class `TestNetworkProxy`\n81776e61 Add missing headers in object store\n56068af2 Splits class `TestMessageProxy`\n45995c88 Splits class `TestKeyManagerProxy`\nf5f0f9f1 Splits class `TestInstanceHaProxy`\n50043daf Splits the `TestBaremetalProxy` class\n290a6cf5 Add limit resource to shared file system\nbba9c5c5 Fix setting initial object/container metadata.\n6a2420db Splits the `TestAcceleratorProxy` class\na2b7c540 Optimize code by moving misc functions to utils\n7f87d2ee Start splitting the `TestImageProxy` class\n4914d8d7 Add user message to shared file system\n67c8c191 Replace the \u0027try except\u0027 block\nd7d6464c Add user group assignment support in identity\n21795acf Add a manila functional check job\nd62a4e60 Fix key generation for caching\nf7f9446d Add \"security_group_ids\" to Port\u0027s query parameters\n88e520f5 Adds storage pools to shared file system\na8c3cb29 Extend has_version function to accept version parameter\n1909df1f Vpn ike policy resource\n3d6aff34 Add common metadata mixin to block storage and clustering\nc1c8dfb7 Make metadata a common mixin\nf3df2095 Introduce QuotaSet in block storage service\nb7ebc731 Introduce QuotaSet in the compute service\n94cb864d Introduce common class for QuotaSet\nfae7ab06 Fix import order in block_storage tests\n2d56b681 Fix import order in cloud tests\nc6494db0 Move tag mixin into the common\nb59cda5c Fix import order in clustering tests\nadcf98c2 Fix import order in orchestration tests\nb86d3a8c Fix import order in identity tests\n8a1d255c Switch identity roles in the cloud layer to proxy\n95cec287 Fix identity role management\n533ba93a Switch identiy.endpoints in the cloud layer\ndf5f987c Add share resource to shared file system\n9a038d8c Switch identity.service in cloud layer to proxy\ne192f215 Switch identity.domains in cloud layer\n54f910f2 Switch identity groups and users in cloud layer\n74fa9871 Add possibility to create subprojects in cloud layer\n72aa4531 Switch project management in cloud to proxy\nccc6478c Temporarily disable nodepool job\nd6c51c53 Add cleanup function for DNS resource\nf5c9653a Switch BS volume_attachment operations in cloud layer\n485b53bd Switch BS backup and snapshot methods in cloud layer to proxy\n734ae40f Add BS snapshot and backup actions\n118cf735 Switch BS type access operations in cloud layer\n8579b095 Add BS type access into proxy/resource layer\nd235b141 Switch block_storage.volume operations in cloud layer\nf9183cf4 Add block_storage.volume actions\nd1c9a96e Copy find_volume to block_storage.v2 proxy\na3582786 Fix active status for block storage\ne6fed6d2 Add scheduler hints on the block_storage.volume resource\n62781ae2 Sort block storage properties alphabetically\nc6177a47 Drop cloud layer methods for Senlin\n7cee92d5 Replace deprecated inspect.getargspec call\n28875355 Revert \"Update TOX_CONSTRAINTS_FILE for feature/r1\"\n698ac14a Switch object_store cloud functions to proxy layer\n70a06d99 Stop sending tenant_id to Neutron\n6db391f6 Switch networking function in cloud layer to proxy\n03b2b3b0 Add access to the resource attribute by server-side name\n2d16931c Update TOX_CONSTRAINTS_FILE for feature/r1\n7a19a5ff Update .gitreview for feature/r1\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Ie3ea1b21df8b20ca357e0457d39e9a411b27e784\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7934e97d79437c4297645cdc49909205fd42397a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7934e97d79437c4297645cdc49909205fd42397a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
