)]}'
{"id":"openstack%2Freleases~841853","triplet_id":"openstack%2Freleases~master~I556b217d7a9030e6c5abb5e9789ed1bc8d8edda0","project":"openstack/releases","branch":"master","topic":"zed-milestone-1","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-05-20 19:37:55.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-05-20 19:37:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I556b217d7a9030e6c5abb5e9789ed1bc8d8edda0","subject":"Release python-swiftclient for Zed-1 milestone","status":"MERGED","created":"2022-05-16 08:17:29.000000000","updated":"2022-05-20 19:37:55.000000000","submitted":"2022-05-20 19:37:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841853-zed-milestone-1","meta_rev_id":"6be4ff670871cb27ad58f618ebc2b9103bb369d7","_number":841853,"virtual_id_number":841853,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-20 19:37:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2022-05-20 19:27:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2022-05-20 19:27:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-16 08:18:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-16 08:24:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-16 11:21:30.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2022-05-20 19:27:18.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"fcf54dd0f7cba364e3199968bfebc2167893d807","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-16 08:17:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60054cf475f393eae06fcd2e0dd61ad2847af2cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 08:24:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/954b95bc89364541a6ed5bb820323445 : SUCCESS in 6m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/942dc7418b5d44738779f1945c075664 : SUCCESS in 4m 59s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/517732487ad74063af62da0686fcfd37 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"c7ea29ff031c6add13eda2e17b17e0082dd71de8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-16 10:19:14.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"adce9a5409f232748d479906a7aba18bf42a52eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 10:27:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fb52095de9549da92d02cc3377b2a77 : SUCCESS in 5m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a2d5db92035b40f6ba175cf5b658e73c : SUCCESS in 5m 52s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c8bace7e44684d3896929874057698f0 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":2},{"id":"5ba567d2fb8322c7238513226ba0d2c4a1f23b4f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-05-20 19:27:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2318e3450fde742a06d1c2fd9d06b658f5e72ee8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 19:28:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"77fb211e325a86488e2437b96a740a771b83ea17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 19:37:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc6a418622e74f6f99b060d7034dd880 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/171fe5ee79494dc19d2e219c1b60ab61 : SUCCESS in 6m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b750c3c802bd4424b2a6d1cbf4c853e0 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":2},{"id":"6be4ff670871cb27ad58f618ebc2b9103bb369d7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 19:37:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8ef04812d8a8dfe997ac3775606f2e302389d570","revisions":{"2287f061d62b52e923db97509db05b185767b11e":{"kind":"REWORK","_number":1,"created":"2022-05-16 08:17:29.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/53/841853/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/53/841853/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/841853/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/841853/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/841853/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/53/841853/1"}}},"commit":{"parents":[{"commit":"c9d01acbb6ee0b5ceaafd387932089df13a3f434","subject":"Merge \"[Puppet OpenStack] Transition stein to EOL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c9d01acbb6ee0b5ceaafd387932089df13a3f434"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-16 08:17:22.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-16 08:17:22.000000000","tz":120},"subject":"Release python-swiftclient for Yoga-1 milestone","message":"Release python-swiftclient for Yoga-1 milestone\n\nThis is a library release for python-swiftclient for the Yoga-1\nmilestone. This repository contains commits that have not been releases\nfor this cycle yet.\n\nIf the team is ready to process with this release, please leave a +1\nto indicate we should go ahead with the release.\n\nIf the team needs more time for things about to merge, or if there is\nsome other reason a release should not be done at this time, please\nleave a -1 with a comment indicating the status. Then update the patch\nwith the new commit hash to use once the team is ready to do the\nrelease.\n\n$ git log --oneline --no-merges 3.13.1..95f68cd\n95f68cd Add timeout for Swift service\n20d837a Remove unnecessary unicode prefixes\n61ce5ac Remove unnecessary object subclassing\nfa137a5 Remove six\n4983b90 Remove coding comments\n3d10744 Remove __future__ imports\nc09621e Don\u0027t patch Requests globally on import\n2636965 Drop support for Python 2\n22a05b2 Add Python3 zed unit tests\n00f5b89 CI: Drop swiftclient-swift-functional-py2 job\nc2d4fc7 Update master for stable/yoga\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I556b217d7a9030e6c5abb5e9789ed1bc8d8edda0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2287f061d62b52e923db97509db05b185767b11e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2287f061d62b52e923db97509db05b185767b11e"}]},"branch":"refs/heads/master"},"8ef04812d8a8dfe997ac3775606f2e302389d570":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-05-16 10:19:14.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/53/841853/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/53/841853/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/841853/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/841853/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/841853/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/53/841853/2"}}},"commit":{"parents":[{"commit":"c9d01acbb6ee0b5ceaafd387932089df13a3f434","subject":"Merge \"[Puppet OpenStack] Transition stein to EOL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c9d01acbb6ee0b5ceaafd387932089df13a3f434"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-16 08:17:22.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2022-05-16 10:19:14.000000000","tz":0},"subject":"Release python-swiftclient for Zed-1 milestone","message":"Release python-swiftclient for Zed-1 milestone\n\nThis is a library release for python-swiftclient for the Zed-1\nmilestone. This repository contains commits that have not been releases\nfor this cycle yet.\n\nIf the team is ready to process with this release, please leave a +1\nto indicate we should go ahead with the release.\n\nIf the team needs more time for things about to merge, or if there is\nsome other reason a release should not be done at this time, please\nleave a -1 with a comment indicating the status. Then update the patch\nwith the new commit hash to use once the team is ready to do the\nrelease.\n\n$ git log --oneline --no-merges 3.13.1..95f68cd\n95f68cd Add timeout for Swift service\n20d837a Remove unnecessary unicode prefixes\n61ce5ac Remove unnecessary object subclassing\nfa137a5 Remove six\n4983b90 Remove coding comments\n3d10744 Remove __future__ imports\nc09621e Don\u0027t patch Requests globally on import\n2636965 Drop support for Python 2\n22a05b2 Add Python3 zed unit tests\n00f5b89 CI: Drop swiftclient-swift-functional-py2 job\nc2d4fc7 Update master for stable/yoga\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I556b217d7a9030e6c5abb5e9789ed1bc8d8edda0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8ef04812d8a8dfe997ac3775606f2e302389d570"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8ef04812d8a8dfe997ac3775606f2e302389d570"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
