)]}'
{"id":"openstack%2Freleases~843623","triplet_id":"openstack%2Freleases~master~Ie0dcddfe271e8fd77a6d40fa54c8ef2ee6914b78","project":"openstack/releases","branch":"master","topic":"slurp-releases","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-12-02 14:37:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie0dcddfe271e8fd77a6d40fa54c8ef2ee6914b78","subject":"Add SLURP mark for Releases page template","status":"MERGED","created":"2022-05-27 14:36:37.000000000","updated":"2022-12-02 14:37:01.000000000","submitted":"2022-12-02 14:37:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"843623-slurp-releases","meta_rev_id":"40a3de3e79f411fe16f68ec54dbc18f29da17968","_number":843623,"virtual_id_number":843623,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-02 14:37:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2022-12-02 14:20:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2022-12-02 14:20:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2022-12-02 14:20:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-27 14:45:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-02 14:20:21.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2022-12-02 14:20:27.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"fb008ab33677fc6c24eda60308d90d0e11be0071","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-05-27 14:36:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a39e4ba7cb3d0333d2be522fefb4a785f468aa9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 14:45:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/418e6f83103e424fb98c747ff8361e27 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ec15c3b071b442da9eebcb6e75b0b15 : SUCCESS in 6m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/65affec4feda4fa095e4ad7856760288 : FAILURE in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"0b71bd16c0432f719eebfb52df075d85ad73b0e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-30 13:08:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d903a9f7ff46809f8a939f70c15bb462306e1b3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 13:19:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6fe3d279eb474c379938c9d8a6ef6668\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ca805242a25493a832b526dbdce9fe6 : SUCCESS in 9m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5c3ef998184742d6b00dc91344e01bbe : FAILURE in 7m 12s","accounts_in_message":[],"_revision_number":2},{"id":"a809b4bc9edc9fa10cd96fe03771d4cd3ad03fdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-30 13:34:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"caf64593be147b668bdc57768c72cda10e217d61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 13:47:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ccdee1a675941c79c947f9d3c9eb10f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f50cbbbb74cc46a78614dee9f945ac15 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e3079bf1ad04f4db03a5f8e634343ca : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d872b9cb15304c52b91dabad0907da3d : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/35b6b7879e3445a79ee291708e7487b6 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7028aad8283e4cdbaac7a6671389cc11 : SUCCESS in 7m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/35558b9bac6c43c6871fbe4121dc2377 : SUCCESS in 4m 45s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c9280a4746f54b98be0c518035266679 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":3},{"id":"6538ca5c07430c8295a71b3e46d76e1bd6699c17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-30 13:57:07.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"28163d27c762c0794edc456b0dfc2821578d5cb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 14:14:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8d5c6c44b0a4319b777dca1971920e9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ea13b7dbd4f4e028ee64c81d5498d42 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/998590206e7f4302a2cdf584f80bbd93 : SUCCESS in 7m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6cbf8c309a0e4c21bc5f038be012fb8e : SUCCESS in 7m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d5cff9f5e5e44f1ab48af772729b9e35 : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e86c69d610c4499aae58a095ec8cf18 : SUCCESS in 8m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa4ab9f862914e0ea8821e1efaa0c827 : SUCCESS in 7m 23s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/433a51630f0b435bb8afa9fe2024737f : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":4},{"id":"4a8c082804c8e00f70ac9d5ab6a1a57a851de791","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-30 15:03:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2b506db47350f650f7aac878c685bf625ddf22da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-12-02 10:47:30.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a02ba33f7d596c31b6507fb13e17394d7007e320","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 10:55:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dc7c66af752480cb738bf96940ff0d8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba9cb9d3dbe1497ca12475f951639ad1 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74ad4b56886d4805b50fd1d1473225e0 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/910ae0b88d7d4fdebba22412e7f39a11 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/584c5b86e965437cbb616d77074c9ee3 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b178905c03b74a86b0c7d8d8ddecc17a : SUCCESS in 6m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/705e9b47c7b0455dac7e151cb96b8102 : SUCCESS in 4m 58s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/337a5e5bfe5e4d23be78ebbc8c493626 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":5},{"id":"6f603ed76fff52fced5e19b7d9c3d54324ecc0b6","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-12-02 14:20:21.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"38b4574175ad3189f7e8e4cba9fc1bd0abab7db9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-12-02 14:20:27.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"15d118fde702fb83d362eaf492e7c23d609ad4e1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-12-02 14:20:37.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"10a585a6d95077f3958538d0edd33b7e8badae1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 14:20:57.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"6c2ae1c7f702141fa0eb67c7786cf4b5592d2d88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 14:37:01.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aecb41824240410a8e4913ad49fad63e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f85bbc088a424d84abd53c535551a672 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/074c73530e6c4ea797f08fe4fac20d01 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/926e966ad9804fe1b0bb319ea91c6650 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d0e3aa056d34c1d8d98551f932de1e1 : SUCCESS in 6m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a28d6da7a2c47ea98d201be83dc3fda : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/48fb6874d76e41f79b088af944165c31 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":5},{"id":"40a3de3e79f411fe16f68ec54dbc18f29da17968","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 14:37:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3306e86b986d8ddad00ce4e64346dfe5a64ebb3b","revisions":{"b73027f214cf44a849bf13fb4ffcd62e8cf6a039":{"kind":"REWORK","_number":1,"created":"2022-05-27 14:36:37.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/23/843623/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/23/843623/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/843623/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/843623/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/843623/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/23/843623/1"}}},"commit":{"parents":[{"commit":"9b9c19c11441609f2cc60d159b912a8317cc698a","subject":"Merge \"New stable releases for Neutron\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9b9c19c11441609f2cc60d159b912a8317cc698a"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-05-24 17:58:32.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-05-27 14:34:49.000000000","tz":120},"subject":"WIP: Add SLURP mark for Releases page template","message":"WIP: Add SLURP mark for Releases page template\n\nThis patch adds to the Releases page template the possibility to\nindicate whether a release is a SLURP release.\n\nChange-Id: Ie0dcddfe271e8fd77a6d40fa54c8ef2ee6914b78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b73027f214cf44a849bf13fb4ffcd62e8cf6a039"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b73027f214cf44a849bf13fb4ffcd62e8cf6a039"}]},"branch":"refs/heads/master"},"5352f4de24cb9483c7af2fa96c8551baea781bed":{"kind":"REWORK","_number":2,"created":"2022-09-30 13:08:37.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/23/843623/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/23/843623/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/843623/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/843623/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/843623/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/23/843623/2"}}},"commit":{"parents":[{"commit":"f1dda04b63586fc804b5eef967dc99691c895c46","subject":"Merge \"Creating stable zed for requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f1dda04b63586fc804b5eef967dc99691c895c46"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-05-24 17:58:32.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-09-30 13:07:53.000000000","tz":120},"subject":"Add SLURP mark for Releases page template","message":"Add SLURP mark for Releases page template\n\nThis patch adds to the Releases page template the possibility to\nindicate whether a release is a SLURP release.\n\nChange-Id: Ie0dcddfe271e8fd77a6d40fa54c8ef2ee6914b78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5352f4de24cb9483c7af2fa96c8551baea781bed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5352f4de24cb9483c7af2fa96c8551baea781bed"}]},"branch":"refs/heads/master"},"6e89b4d581f28f00e076b6e6643dfe10d1378f72":{"kind":"REWORK","_number":3,"created":"2022-09-30 13:34:47.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/23/843623/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/23/843623/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/843623/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/843623/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/843623/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/23/843623/3"}}},"commit":{"parents":[{"commit":"f1dda04b63586fc804b5eef967dc99691c895c46","subject":"Merge \"Creating stable zed for requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f1dda04b63586fc804b5eef967dc99691c895c46"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-05-24 17:58:32.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-09-30 13:31:12.000000000","tz":120},"subject":"Add SLURP mark for Releases page template","message":"Add SLURP mark for Releases page template\n\nThis patch adds to the Releases page template the possibility to\nindicate whether a release is a SLURP release.\n\nChange-Id: Ie0dcddfe271e8fd77a6d40fa54c8ef2ee6914b78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6e89b4d581f28f00e076b6e6643dfe10d1378f72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6e89b4d581f28f00e076b6e6643dfe10d1378f72"}]},"branch":"refs/heads/master"},"a3cd756cf3b5be0150a617326e8464e5ae611188":{"kind":"REWORK","_number":4,"created":"2022-09-30 13:57:07.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/23/843623/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/23/843623/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/843623/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/843623/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/843623/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/23/843623/4"}}},"commit":{"parents":[{"commit":"f1dda04b63586fc804b5eef967dc99691c895c46","subject":"Merge \"Creating stable zed for requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f1dda04b63586fc804b5eef967dc99691c895c46"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-05-24 17:58:32.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-09-30 13:56:37.000000000","tz":120},"subject":"Add SLURP mark for Releases page template","message":"Add SLURP mark for Releases page template\n\nThis patch adds to the Releases page template the possibility to\nindicate whether a release is a SLURP release.\n\nChange-Id: Ie0dcddfe271e8fd77a6d40fa54c8ef2ee6914b78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a3cd756cf3b5be0150a617326e8464e5ae611188"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a3cd756cf3b5be0150a617326e8464e5ae611188"}]},"branch":"refs/heads/master"},"3306e86b986d8ddad00ce4e64346dfe5a64ebb3b":{"kind":"REWORK","_number":5,"created":"2022-12-02 10:47:30.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/23/843623/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/23/843623/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/843623/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/843623/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/843623/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/23/843623/5"}}},"commit":{"parents":[{"commit":"d62f935a42ba9f7127e0e405c4fd7e3be0699052","subject":"Merge \"[xena] [barbican] [security] Release 13.0.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d62f935a42ba9f7127e0e405c4fd7e3be0699052"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-05-24 17:58:32.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-12-02 10:46:43.000000000","tz":60},"subject":"Add SLURP mark for Releases page template","message":"Add SLURP mark for Releases page template\n\nThis patch adds to the Releases page template the possibility to\nindicate whether a release is a SLURP release.\n\nChange-Id: Ie0dcddfe271e8fd77a6d40fa54c8ef2ee6914b78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3306e86b986d8ddad00ce4e64346dfe5a64ebb3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3306e86b986d8ddad00ce4e64346dfe5a64ebb3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[]}
