)]}'
{"id":"openstack%2Freleases~847180","triplet_id":"openstack%2Freleases~master~I4df60f91427af12926e2043efa2c4a19dbbfbe1e","project":"openstack/releases","branch":"master","topic":"stadium_eol_qrs","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-07-08 12:34:56.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-11-17 10:09:54.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4df60f91427af12926e2043efa2c4a19dbbfbe1e","subject":"Add yaml files for tap-as-a-service","status":"ABANDONED","created":"2022-06-22 15:18:57.000000000","updated":"2022-11-17 10:09:54.000000000","total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"1fbd586d202c44c155590b46bfafaba92c40245f","_number":847180,"virtual_id_number":847180,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-07-08 13:01:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-07-08 12:50:18.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-22 15:20:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-30 12:09:19.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"69a56daf71fdda65cd219f06d9eb5a11fe99853f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-06-22 15:18:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d588d14dc10a270193f093e9f2864cee4a347285","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 15:20:35.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"8a8b4c87fa7d395dde252587208e8e5c9787ccf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 15:30:50.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd0bd56eefc3406cbf74c215dfdbd206 : SUCCESS in 8m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8a7d8b0682224db983e35330b7dfd868 : FAILURE in 4m 11s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ec9ff5071c0f4ce384bad996bdbc24e4 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"fb4fcb4584e2be26df2d722a206cb41d26530e46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-06-23 06:53:27.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e934f4285b5ac44e038889653f4327cf73b78d37","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-06-23 06:54:06.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"89200770a8a4633f76837191967963e56e0fcd1f","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 06:54:54.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"ac816df8483000072a4c2779b3a53a02fa897bbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 07:12:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db97a644acae434e935a27ec13c569a5 : SUCCESS in 5m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/24d9aeaac9a74efb995b1fe69f8c2260 : SUCCESS in 4m 51s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/cba6d729f0be4784aaa733dabe0af677 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":2},{"id":"1227c758be27d289e7db24712d4d99ad75708ea3","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-01 16:10:53.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"efcd5d9283e5afc01f2c6e6aba555785781a9f49","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-07-04 12:18:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"30ffe2cbd8b8cb94eea4f85e0d400783e21608d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-07-08 09:25:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"df7921bb8cac40c882058e3cc9cd4465ed01dcd5","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 09:26:01.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"a6ba35f595b31a8b68a1d25d93d82a340d187e37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 09:35:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/77fa1931aee6443eb24f70e9035184fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a044841c08546e391db9eaa97c2356d : SUCCESS in 6m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5e3b2385fa284ef0b0976051550cf0f9 : FAILURE in 4m 46s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/dfcccaba6eab4a1e8fdb5026d090f259 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":3},{"id":"9dcb886136a09291d8d43d977a5cb40cbf25f2a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-07-08 09:42:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2a4978c7a718038c280d4294c3c47b6d32c210e3","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 09:43:43.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"87c96d51779c51095ddefeb937488af319143647","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 09:54:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a646d0dac6a4960940d381abf5500f5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdae49568f1248ecab07a83db7dccf3d : SUCCESS in 6m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6b8e96fd37e245639226d8cedb1a24e4 : FAILURE in 4m 18s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a479d96c2b3c41c997a950bd2ea80a1a : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":4},{"id":"8edea174a8362c52be16d9a8f007c889785966d2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-08 12:34:56.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7c82f1b63689aed13cc3e62405610fed71e6996a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-07-08 12:49:25.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"47113db1999d83ba922ef6808c81986621d99f95","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 12:50:18.000000000","message":"Patch Set 5: PTL-Approved+1","accounts_in_message":[],"_revision_number":5},{"id":"b2d6faae009472bc419401f839bebe5b842d32cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 13:01:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/19c7a995c7a9484db250ae2e80542285\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfb1926e72914bb88d15c3f16a59d9bd : SUCCESS in 7m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9675c0285e9c41aaa15c2192c4c4f293 : FAILURE in 5m 58s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f7f8b24c4f64408592909a5365bf5025 : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":5},{"id":"1fbd586d202c44c155590b46bfafaba92c40245f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-11-17 10:09:54.000000000","message":"Abandoned\n\nas it would cause more trouble let\u0027s forget it","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c813cca4f86b82ca66955ed5a14cba3fccd4e611","revisions":{"76ec53dacc9beaa75d04c073aaacc1dfcf24231d":{"kind":"REWORK","_number":1,"created":"2022-06-22 15:18:57.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/80/847180/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/80/847180/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/847180/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/847180/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/847180/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/80/847180/1"}}},"commit":{"parents":[{"commit":"e0dbe6219b3bbccdc17058eb890451f6601397b3","subject":"Release taskflow 5.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e0dbe6219b3bbccdc17058eb890451f6601397b3"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-06-22 15:13:56.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-06-22 15:13:56.000000000","tz":120},"subject":"Add yaml files for tap-as-a-service","message":"Add yaml files for tap-as-a-service\n\nAs tap-as-a-service was under x/ namespace for several cycles we miss\nyaml files for some of its branches and tags.\nThis patch adds these files for\n* queens (with EOL tag), tag 3.0.0\n* rocky  (with EOL tag), tag 4.0.0\n* stein  (with EOL tag), tag 5.0.0\n* train  tag 6.0.0\n* victoria tag 7.0.0\n* xena tag 8.0.0\n\nWith these yaml files in place the later management of taas versions and\ntags  will be easier I hope.\n\nChange-Id: I4df60f91427af12926e2043efa2c4a19dbbfbe1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/76ec53dacc9beaa75d04c073aaacc1dfcf24231d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/76ec53dacc9beaa75d04c073aaacc1dfcf24231d"}]},"branch":"refs/heads/master"},"c229e550136af6ad70b6132984e03e3d99bb2dd4":{"kind":"REWORK","_number":2,"created":"2022-06-23 06:53:27.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/80/847180/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/80/847180/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/847180/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/847180/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/847180/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/80/847180/2"}}},"commit":{"parents":[{"commit":"e0dbe6219b3bbccdc17058eb890451f6601397b3","subject":"Release taskflow 5.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e0dbe6219b3bbccdc17058eb890451f6601397b3"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-06-22 15:13:56.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2022-06-23 06:53:27.000000000","tz":0},"subject":"Add yaml files for tap-as-a-service","message":"Add yaml files for tap-as-a-service\n\nAs tap-as-a-service was under x/ namespace for several cycles we miss\nyaml files for some of its branches and tags.\nThis patch adds these files for\n* queens (with EOL tag), tag 3.0.0\n* rocky  (with EOL tag), tag 4.0.0\n* stein  (with EOL tag), tag 5.0.0\n* train  tag 6.0.0\n* victoria tag 7.0.0\n* xena tag 8.0.0\n\nWith these yaml files in place the later management of taas versions and\ntags  will be easier I hope.\n\nChange-Id: I4df60f91427af12926e2043efa2c4a19dbbfbe1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c229e550136af6ad70b6132984e03e3d99bb2dd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c229e550136af6ad70b6132984e03e3d99bb2dd4"}]},"branch":"refs/heads/master"},"ed3076d6ab6bf3df0dba69f6ebad538da39f7a1c":{"kind":"REWORK","_number":3,"created":"2022-07-08 09:25:11.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/80/847180/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/80/847180/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/847180/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/847180/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/847180/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/80/847180/3"}}},"commit":{"parents":[{"commit":"e0dbe6219b3bbccdc17058eb890451f6601397b3","subject":"Release taskflow 5.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e0dbe6219b3bbccdc17058eb890451f6601397b3"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-06-22 15:13:56.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-07-08 09:24:21.000000000","tz":120},"subject":"Add yaml files for tap-as-a-service","message":"Add yaml files for tap-as-a-service\n\nAs tap-as-a-service was under x/ namespace for several cycles we miss\nyaml files for some of its branches and tags.\nThis patch adds these files for\n* queens (with EOL tag)\n* rocky  (with EOL tag)\n* stein  (with EOL tag)\n\nWith these yaml files in place the later management of taas versions and\ntags  will be easier I hope.\n\nChange-Id: I4df60f91427af12926e2043efa2c4a19dbbfbe1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ed3076d6ab6bf3df0dba69f6ebad538da39f7a1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ed3076d6ab6bf3df0dba69f6ebad538da39f7a1c"}]},"branch":"refs/heads/master"},"d504074017c008884aa7134ecbc5b0bd233150e2":{"kind":"REWORK","_number":4,"created":"2022-07-08 09:42:45.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/80/847180/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/80/847180/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/847180/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/847180/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/847180/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/80/847180/4"}}},"commit":{"parents":[{"commit":"e0dbe6219b3bbccdc17058eb890451f6601397b3","subject":"Release taskflow 5.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e0dbe6219b3bbccdc17058eb890451f6601397b3"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-06-22 15:13:56.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-07-08 09:42:34.000000000","tz":120},"subject":"Add yaml files for tap-as-a-service","message":"Add yaml files for tap-as-a-service\n\nAs tap-as-a-service was under x/ namespace for several cycles we miss\nyaml files for some of its branches and tags.\nThis patch adds these files for\n* queens (with EOL tag)\n* rocky  (with EOL tag)\n* stein  (with EOL tag)\n\nWith these yaml files in place the later management of taas versions and\ntags  will be easier I hope.\n\nChange-Id: I4df60f91427af12926e2043efa2c4a19dbbfbe1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d504074017c008884aa7134ecbc5b0bd233150e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d504074017c008884aa7134ecbc5b0bd233150e2"}]},"branch":"refs/heads/master"},"c813cca4f86b82ca66955ed5a14cba3fccd4e611":{"kind":"REWORK","_number":5,"created":"2022-07-08 12:49:25.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/80/847180/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/80/847180/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/847180/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/847180/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/847180/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/80/847180/5"}}},"commit":{"parents":[{"commit":"e0dbe6219b3bbccdc17058eb890451f6601397b3","subject":"Release taskflow 5.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e0dbe6219b3bbccdc17058eb890451f6601397b3"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-06-22 15:13:56.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2022-07-08 12:49:25.000000000","tz":0},"subject":"Add yaml files for tap-as-a-service","message":"Add yaml files for tap-as-a-service\n\nAs tap-as-a-service was under x/ namespace for several cycles we miss\nyaml files for some of its branches and tags.\nThis patch adds these files for\n* queens (with EOL tag)\n* rocky  (with EOL tag)\n* stein  (with EOL tag)\n\nWith these yaml files in place the later management of taas versions and\ntags  will be easier I hope.\n\nChange-Id: I4df60f91427af12926e2043efa2c4a19dbbfbe1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c813cca4f86b82ca66955ed5a14cba3fccd4e611"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c813cca4f86b82ca66955ed5a14cba3fccd4e611"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
