)]}'
{"id":"openstack%2Freleases~848842","triplet_id":"openstack%2Freleases~master~I7d3501a9cd2453e545826d158223dcd1bddeda7a","project":"openstack/releases","branch":"master","topic":"wallaby-em","attention_set":{},"removed_from_attention_set":{"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2022-11-03 11:16:49.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-11-02 09:01:49.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-11-03 11:16:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7d3501a9cd2453e545826d158223dcd1bddeda7a","subject":"Cinder team releases for stable/wallaby","status":"MERGED","created":"2022-07-06 13:05:16.000000000","updated":"2022-11-03 11:16:49.000000000","submitted":"2022-11-03 11:16:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848842-wallaby-em","meta_rev_id":"bfaf1691020d9b426edbdab2625b359d46e92a81","_number":848842,"virtual_id_number":848842,"owner":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-03 11:16:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2022-11-02 09:01:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2022-11-03 11:00:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-01 19:08:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2022-11-03 11:00:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-11-01 18:17:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-06 13:06:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-06 15:05:56.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2022-07-06 15:05:56.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2022-11-03 11:00:57.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"ddc58f746d1980d2c527f68d12038b05d6b27667","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2022-07-06 13:05:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"897b49bb2f22774b1c59002c8195d262065571c5","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 13:06:30.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"644cf4178e89bdaa7f64a9870dfc5d290490d54c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 13:23:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2986f28ebe954049991d1874845c7ecf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e3ab73459c3402499489949ba1c9d50 : SUCCESS in 6m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6b677cd13d584e26adc41083e158af49 : SUCCESS in 16m 17s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1fc1bc6f9fc0415983d2db8e1186f262 : SUCCESS in 7m 33s","accounts_in_message":[],"_revision_number":1},{"id":"f3c46f775dbc12192ccd305a66f4410cd8107c44","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-06 20:18:28.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c0e604c3d315f833d940594b7321f2f4ce4a5c6e","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-10-28 09:06:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d1ca7969866c54feb196ae17edd02765a1096433","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 09:07:09.000000000","message":"Patch Set 1: -PTL-Approved","accounts_in_message":[],"_revision_number":1},{"id":"36bf7ce745253d9954e0943add523c66dab4c397","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2022-11-01 16:10:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"beaf568bf7e92e1fd1ac8865477c0eeea5ee34cf","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-01 16:10:58.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"883cb18db3a3c6762959d4b98e9878c614d94120","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2022-11-01 16:11:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e882a7af71b27973f022aba11893d4fcf6376468","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-01 16:25:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2465ea8fb7fe4824b5532a3e28980667\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd0849ac262241389f937d51d4d3e302 : SUCCESS in 7m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e4f40da71a704e9ab25b1bbd56a79696 : SUCCESS in 13m 51s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/52f76724cdbc4d5c85e96dfa794722e7 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":2},{"id":"a7205edfbbeac8050cd9ac9d060d4a15d725052b","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-11-01 17:24:08.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"812f8923b0d8a6ec7af1a30dec38aed5d69bea73","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-11-01 17:24:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"54130272cd8a4cea0c0b2add983ee16940ddca3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2022-11-01 18:16:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c6f501feef5e353b8ebcc924cd0cd6eceaeedfdb","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-01 18:17:35.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"ce9a2012a02caa642b41df48d5c8ee58cc3fa4de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-01 18:39:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/738f2693a76a434cba0ec69165015089\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/804954cf34084791badea8b0c4acbc3b : SUCCESS in 6m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1f35126860f64da68315bf4e265cdbde : SUCCESS in 21m 32s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/17dd3b0b6b674af3b4b08b6f3b880ccf : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":3},{"id":"b571b3f4d7b75e52289299111242a90f1140b96e","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-11-01 19:08:51.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ab73433c8f0aba421dba64efc2ab5d7c9668b151","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-11-02 09:01:49.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b27ab66cbb503914649555465a07fea89355fc16","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-11-02 15:35:08.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2a27138b040cd7bd3799a09bf1a8b21015ceaeea","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-03 11:00:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6368490cee789c7515adace4182fbbdc466d4d7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 11:01:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d89482f412e3573f6169dc38c7b4d2ac338d4025","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 11:16:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbbf1e4455c64eaeb4a6d6024adaaa7d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bde113d51cd47e2b39e35555c4b30f8 : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4270acef46bc418191b3a3a6db26a81f : SUCCESS in 5m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/de6555de4fa9452b9456e4866a767c46 : SUCCESS in 14m 45s","accounts_in_message":[],"_revision_number":3},{"id":"bfaf1691020d9b426edbdab2625b359d46e92a81","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 11:16:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5aa93edc2aec2fbbb7eda63e691dd0bcfb608258","revisions":{"bf2fa71ec686db2b67cf92c5289e41b63a82fbb5":{"kind":"REWORK","_number":1,"created":"2022-07-06 13:05:16.000000000","uploader":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"ref":"refs/changes/42/848842/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/42/848842/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/848842/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/848842/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/848842/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/42/848842/1"}}},"commit":{"parents":[{"commit":"fd13216ae4ee0c62acea7471a4ccd8b3457e51c5","subject":"Cinder team releases for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fd13216ae4ee0c62acea7471a4ccd8b3457e51c5"}]}],"author":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2022-07-06 00:55:10.000000000","tz":0},"committer":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2022-07-06 13:00:20.000000000","tz":0},"subject":"Cinder team releases for stable/wallaby","message":"Cinder team releases for stable/wallaby\n\nNot being released at this time:\n- python-brick-cinderclient-ext (no functional changes)\n- python-cinderclient (no functional changes)\n\nChange-Id: I7d3501a9cd2453e545826d158223dcd1bddeda7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bf2fa71ec686db2b67cf92c5289e41b63a82fbb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bf2fa71ec686db2b67cf92c5289e41b63a82fbb5"}]},"branch":"refs/heads/master"},"83c916531f0e6b51a5e73c44438cb137eb6b3c56":{"kind":"REWORK","_number":2,"created":"2022-11-01 16:10:09.000000000","uploader":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"ref":"refs/changes/42/848842/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/42/848842/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/848842/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/848842/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/848842/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/42/848842/2"}}},"commit":{"parents":[{"commit":"d6031738ef67a6d7741f353413e1dd63b28e6f4c","subject":"Merge \"[zaqar] Transition Wallaby to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6031738ef67a6d7741f353413e1dd63b28e6f4c"}]}],"author":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2022-07-06 00:55:10.000000000","tz":0},"committer":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2022-11-01 16:09:18.000000000","tz":-240},"subject":"Cinder team releases for stable/wallaby","message":"Cinder team releases for stable/wallaby\n\nNot being released at this time:\n- cinderlib (no functional changes)\n- python-brick-cinderclient-ext (no functional changes)\n- python-cinderclient (no functional changes)\n\nChange-Id: I7d3501a9cd2453e545826d158223dcd1bddeda7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/83c916531f0e6b51a5e73c44438cb137eb6b3c56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/83c916531f0e6b51a5e73c44438cb137eb6b3c56"}]},"branch":"refs/heads/master"},"5aa93edc2aec2fbbb7eda63e691dd0bcfb608258":{"kind":"REWORK","_number":3,"created":"2022-11-01 18:16:44.000000000","uploader":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"ref":"refs/changes/42/848842/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/42/848842/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/848842/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/848842/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/848842/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/42/848842/3"}}},"commit":{"parents":[{"commit":"d6031738ef67a6d7741f353413e1dd63b28e6f4c","subject":"Merge \"[zaqar] Transition Wallaby to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6031738ef67a6d7741f353413e1dd63b28e6f4c"}]}],"author":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2022-07-06 00:55:10.000000000","tz":0},"committer":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2022-11-01 18:15:08.000000000","tz":-240},"subject":"Cinder team releases for stable/wallaby","message":"Cinder team releases for stable/wallaby\n\nNot being released at this time:\n- cinderlib (no functional changes)\n- python-brick-cinderclient-ext (no functional changes)\n- python-cinderclient (no functional changes)\n\nChange-Id: I7d3501a9cd2453e545826d158223dcd1bddeda7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5aa93edc2aec2fbbb7eda63e691dd0bcfb608258"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5aa93edc2aec2fbbb7eda63e691dd0bcfb608258"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[]}
