)]}'
{"id":"openstack%2Freleases~849994","triplet_id":"openstack%2Freleases~master~Ie5d9eda9581471e754ffda4026adde7eade5226c","project":"openstack/releases","branch":"master","topic":"vf/1.7.1","attention_set":{},"removed_from_attention_set":{"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2022-08-04 10:09:03.000000000","reason":"Change was submitted"},"1955":{"account":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"last_update":"2022-08-04 10:09:03.000000000","reason":"Change was submitted"},"16515":{"account":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"last_update":"2022-08-04 10:09:03.000000000","reason":"Change was submitted"},"35059":{"account":{"_account_id":35059,"name":"Veronika Fisarova","display_name":"vfisarov","email":"vfisarov@redhat.com","username":"vfisarov"},"last_update":"2022-08-04 10:09:03.000000000","reason":"Change was submitted"},"27427":{"account":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"last_update":"2022-07-19 12:51:19.000000000","reason":"\u003cGERRIT_ACCOUNT_27427\u003e replied on the change","reason_account":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-08-04 08:45:47.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"15895":{"account":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"last_update":"2022-08-04 10:09:03.000000000","reason":"Change was submitted"},"7144":{"account":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"last_update":"2022-08-04 10:09:03.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-08-04 10:09:03.000000000","reason":"Change was submitted"},"32926":{"account":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"last_update":"2022-08-04 10:09:03.000000000","reason":"Change was submitted"},"30750":{"account":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"last_update":"2022-08-04 10:09:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie5d9eda9581471e754ffda4026adde7eade5226c","subject":"Push 1.7.1 release for Validations Framework","status":"MERGED","created":"2022-07-15 11:55:33.000000000","updated":"2022-08-04 10:09:03.000000000","submitted":"2022-08-04 10:09:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849994-vf/1.7.1","meta_rev_id":"e0478357e2d065ec9d6f71ab0564eabb99a54e5a","_number":849994,"virtual_id_number":849994,"owner":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-04 10:09:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"recommended":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":1,"date":"2022-08-04 07:32:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":2,"date":"2022-08-04 09:54:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2022-08-04 08:45:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":1,"date":"2022-08-04 09:54:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-08-04 07:33:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":35059,"name":"Veronika Fisarova","display_name":"vfisarov","email":"vfisarov@redhat.com","username":"vfisarov"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},{"_account_id":35059,"name":"Veronika Fisarova","display_name":"vfisarov","email":"vfisarov@redhat.com","username":"vfisarov"}],"CC":[{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-15 11:56:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-15 12:03:31.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"state":"REVIEWER"},{"updated":"2022-07-15 12:03:31.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":35059,"name":"Veronika Fisarova","display_name":"vfisarov","email":"vfisarov@redhat.com","username":"vfisarov"},"state":"REVIEWER"},{"updated":"2022-07-15 12:03:31.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"state":"REVIEWER"},{"updated":"2022-07-15 12:03:31.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2022-07-15 12:05:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-15 12:25:49.000000000","updated_by":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2022-07-18 09:23:53.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"state":"REVIEWER"},{"updated":"2022-07-18 15:20:22.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2022-07-20 07:31:29.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2022-07-20 07:31:29.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-07-25 14:29:33.000000000","updated_by":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"reviewer":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"state":"CC"},{"updated":"2022-08-04 09:54:49.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"d678146fab36c209dce11347cf0ab9eefda6a068","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2022-07-15 11:55:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81e534a88e44b6312f1ff22868d2b08ae432c62a","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-07-15 12:03:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"005d5d46054c910539affd8a0de153876cffacbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 12:05:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdfbe68d6d274fcf923359f99d81de52\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f933969e2ed4598902be3abcad38317 : SUCCESS in 9m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/91834115ccc84c7ab8242395555b79be : SUCCESS in 8m 45s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0ccd081ab7e04558a3686d4c0123ca3a : SUCCESS in 7m 46s","accounts_in_message":[],"_revision_number":1},{"id":"8c8c5c1cac666e0d01e1af2369d807ec16c62feb","author":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"date":"2022-07-19 12:51:19.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"43bfbc55348fa99206b2455085edfea7b644f9ba","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-25 11:58:02.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f864c76508ad1656e8704bdde6eed56179f15978","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-07-25 12:15:46.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"694de0b6939f81e186c9ab2f39a1fc216e68c55e","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-07-25 12:15:59.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"71ab4115d1895acb97b69296a8ba42930184eaa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 12:27:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/47005b7b78504b1c897e3c1017e56149\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6f31617b2724d498b12a32cd049a17a : FAILURE in 3m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d56efaaafb3940489c0e54833dffbd17 : FAILURE in 6m 42s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/49340c406f984543a964b9a1f2d006a6 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":2},{"id":"894b89d12380a0174ca254febcfb9a7cbffdac28","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-25 12:32:55.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7c6fa0c272cd64c20952554254ed762a7a6f6bcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-07-25 12:38:41.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"7452d0e64fbb3ac9feb22430793fcd9131d7901b","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-07-25 12:38:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7e5f0606b9590b2655c797457f47760a020d0a57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-07-25 12:39:41.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"afcfa2c16b5499a87cdf8aa133311cc25cd94cdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-07-25 12:41:31.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"ae3c14ec96497f14869cd35c59236b8ee4e0aede","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 12:53:36.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e45fb7db4a014d5895627aac2cbe06a5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b060d81d7d5a4229a79ec01037d0fc94 : FAILURE in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a58cc88f569e45ee806bf4b9f2114f62 : SUCCESS in 6m 27s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0d956cc9ef91485da59af103ec85c638 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":5},{"id":"d4aabf1587ffe072ca2bd43279a45ff58539c7cd","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-07-25 14:09:48.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cd4ae0e81a37635b6e5f7e65d626a94b9974b034","author":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"date":"2022-07-25 14:29:33.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f9b15eab177ec93808d1070438160e894af3eace","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-07-26 07:29:33.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"4c7ebe3642f4b9fabc4a81c070cf5d747437dae9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 07:41:05.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d173ada8e54b4e57b58ea3afdbd48cba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5532349d43b04a3f842809d584496e62 : SUCCESS in 9m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3fc7114811a345f081b677cd51807180 : SUCCESS in 6m 29s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/eae9802d5f8b428d83f80d0699fa699e : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":6},{"id":"5b48f0ddbc18bdf92e3cf896c29bc6cb66524510","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-07-27 08:36:03.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"fd1e286680cc28686e804f1ca99446f93bdab6dd","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-08-01 08:57:25.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"c0c8c78bc779df6d9c44afac8e794d5e07d3895a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-08-01 09:00:51.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"8b62294505770d5d21b33355da408902bd593e25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-08-01 09:00:55.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"81869ff39254ec5e4629f44ebef3c34d386ab389","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 09:12:21.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4090cd90f4424370ae4bc871a0936d54\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b6b6e1727fc4c65a2cab6be609b74c8 : SUCCESS in 7m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1c2b6cc882ea4783bef26c9bfb84166d : SUCCESS in 9m 08s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/982c85bf943c476380be9810333f7c86 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":8},{"id":"548be3e2ad0f5eead579ddb4e8ce102ed1ffb9ed","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-08-04 07:32:35.000000000","message":"Patch Set 8: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"1c1c0500c141badba5450a73e0dd3406ad94e13b","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 07:33:31.000000000","message":"Patch Set 8: PTL-Approved+1","accounts_in_message":[],"_revision_number":8},{"id":"bc86ab75af71159eb9b4e1d030e3a9f09ce16f6d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-08-04 08:45:47.000000000","message":"Patch Set 8: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"b39284c31a9a05ae4572720fa92d79e5ad33eb0c","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-08-04 09:54:49.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"68e646ebeb6cf6389fe64089ed124ba20b06f938","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 09:55:20.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"c304db93f71e0af78c35a3822b0c9f43ccdfa84c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 10:09:03.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ecc9b3d506e466386cb3b37579edf04\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45ea23f57c264ebe9446e3aa550e09c8 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc7a96e8bed045a0ae59a3b87c2eb0fd : SUCCESS in 7m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4351b570d9c942aba8a308707a21b615 : SUCCESS in 7m 15s","accounts_in_message":[],"_revision_number":8},{"id":"e0478357e2d065ec9d6f71ab0564eabb99a54e5a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 10:09:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"536378904f2fa4801e74a7bb07ab2f47b474942c","revisions":{"238e2b646fdf7c64502f796028003fb5d37de1cf":{"kind":"REWORK","_number":1,"created":"2022-07-15 11:55:33.000000000","uploader":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"ref":"refs/changes/94/849994/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/94/849994/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/849994/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/849994/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/849994/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/94/849994/1"}}},"commit":{"parents":[{"commit":"d7e145811baaed775e57c08d57d81a34f5822d0c","subject":"Merge \"Zed-2 Release for metalsmith\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d7e145811baaed775e57c08d57d81a34f5822d0c"}]}],"author":{"name":"matbu","email":"mat.bultel@gmail.com","date":"2022-07-15 11:54:48.000000000","tz":120},"committer":{"name":"matbu","email":"mat.bultel@gmail.com","date":"2022-07-15 11:54:48.000000000","tz":120},"subject":"Push 1.7.1 release for Validations Framework","message":"Push 1.7.1 release for Validations Framework\n\nPushed 1.7.1 release for validations framework:\n- validations-libs\n- validations-common\n\nChange-Id: Ie5d9eda9581471e754ffda4026adde7eade5226c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/238e2b646fdf7c64502f796028003fb5d37de1cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/238e2b646fdf7c64502f796028003fb5d37de1cf"}]},"branch":"refs/heads/master"},"5ce946e094b1413f46ae34807b232da15dfc246f":{"kind":"REWORK","_number":2,"created":"2022-07-25 12:15:46.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/94/849994/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/94/849994/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/849994/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/849994/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/849994/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/94/849994/2"}}},"commit":{"parents":[{"commit":"d7e145811baaed775e57c08d57d81a34f5822d0c","subject":"Merge \"Zed-2 Release for metalsmith\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d7e145811baaed775e57c08d57d81a34f5822d0c"}]}],"author":{"name":"matbu","email":"mat.bultel@gmail.com","date":"2022-07-15 11:54:48.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-07-25 12:15:45.000000000","tz":0},"subject":"Push 1.7.1 release for Validations Framework","message":"Push 1.7.1 release for Validations Framework\n\nPushed 1.7.1 release for validations framework:\n- validations-libs\n- validations-common\n\nChange-Id: Ie5d9eda9581471e754ffda4026adde7eade5226c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5ce946e094b1413f46ae34807b232da15dfc246f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5ce946e094b1413f46ae34807b232da15dfc246f"}]},"branch":"refs/heads/master"},"b39d55ee3f9ced87dc3f5fb6bb8f92debdefaf58":{"kind":"REWORK","_number":3,"created":"2022-07-25 12:38:41.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/94/849994/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/94/849994/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/849994/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/849994/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/849994/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/94/849994/3"}}},"commit":{"parents":[{"commit":"d7e145811baaed775e57c08d57d81a34f5822d0c","subject":"Merge \"Zed-2 Release for metalsmith\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d7e145811baaed775e57c08d57d81a34f5822d0c"}]}],"author":{"name":"matbu","email":"mat.bultel@gmail.com","date":"2022-07-15 11:54:48.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-07-25 12:38:41.000000000","tz":0},"subject":"Push 1.7.1 release for Validations Framework","message":"Push 1.7.1 release for Validations Framework\n\nPushed 1.7.1 release for validations framework:\n- validations-libs\n- validations-common\n\nChange-Id: Ie5d9eda9581471e754ffda4026adde7eade5226c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b39d55ee3f9ced87dc3f5fb6bb8f92debdefaf58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b39d55ee3f9ced87dc3f5fb6bb8f92debdefaf58"}]},"branch":"refs/heads/master"},"95318537017d1f106a7737aecb009738c4ed84a1":{"kind":"REWORK","_number":4,"created":"2022-07-25 12:39:41.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/94/849994/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/94/849994/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/849994/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/849994/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/849994/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/94/849994/4"}}},"commit":{"parents":[{"commit":"d7e145811baaed775e57c08d57d81a34f5822d0c","subject":"Merge \"Zed-2 Release for metalsmith\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d7e145811baaed775e57c08d57d81a34f5822d0c"}]}],"author":{"name":"matbu","email":"mat.bultel@gmail.com","date":"2022-07-15 11:54:48.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-07-25 12:39:40.000000000","tz":0},"subject":"Push 1.7.1 release for Validations Framework","message":"Push 1.7.1 release for Validations Framework\n\nPushed 1.7.1 release for validations framework:\n- validations-libs\n- validations-common\n\nChange-Id: Ie5d9eda9581471e754ffda4026adde7eade5226c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/95318537017d1f106a7737aecb009738c4ed84a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/95318537017d1f106a7737aecb009738c4ed84a1"}]},"branch":"refs/heads/master"},"94ebff20f1d304f786f881b022d6ee7c6af48608":{"kind":"REWORK","_number":5,"created":"2022-07-25 12:41:31.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/94/849994/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/94/849994/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/849994/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/849994/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/849994/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/94/849994/5"}}},"commit":{"parents":[{"commit":"d7e145811baaed775e57c08d57d81a34f5822d0c","subject":"Merge \"Zed-2 Release for metalsmith\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d7e145811baaed775e57c08d57d81a34f5822d0c"}]}],"author":{"name":"matbu","email":"mat.bultel@gmail.com","date":"2022-07-15 11:54:48.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-07-25 12:41:30.000000000","tz":0},"subject":"Push 1.7.1 release for Validations Framework","message":"Push 1.7.1 release for Validations Framework\n\nPushed 1.7.1 release for validations framework:\n- validations-libs\n- validations-common\n\nChange-Id: Ie5d9eda9581471e754ffda4026adde7eade5226c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/94ebff20f1d304f786f881b022d6ee7c6af48608"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/94ebff20f1d304f786f881b022d6ee7c6af48608"}]},"branch":"refs/heads/master"},"7cddc2da83f61276a7857422e0d73c3d4108f459":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2022-07-26 07:29:33.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/94/849994/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/94/849994/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/849994/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/849994/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/849994/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/94/849994/6"}}},"commit":{"parents":[{"commit":"d7e145811baaed775e57c08d57d81a34f5822d0c","subject":"Merge \"Zed-2 Release for metalsmith\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d7e145811baaed775e57c08d57d81a34f5822d0c"}]}],"author":{"name":"matbu","email":"mat.bultel@gmail.com","date":"2022-07-15 11:54:48.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-07-26 07:29:33.000000000","tz":0},"subject":"Push 1.7.1 release for Validations Framework","message":"Push 1.7.1 release for Validations Framework\n\nPushed 1.7.1 release for validations framework:\n- validations-libs\n- validations-common\n\nDepends-On: https://review.opendev.org/c/openstack/releases/+/850889\n\nChange-Id: Ie5d9eda9581471e754ffda4026adde7eade5226c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7cddc2da83f61276a7857422e0d73c3d4108f459"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7cddc2da83f61276a7857422e0d73c3d4108f459"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8048994bd536fd08b803e4636fff7e90a7de37ec":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2022-08-01 09:00:51.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/94/849994/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/94/849994/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/849994/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/849994/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/849994/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/94/849994/7"}}},"commit":{"parents":[{"commit":"d7e145811baaed775e57c08d57d81a34f5822d0c","subject":"Merge \"Zed-2 Release for metalsmith\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d7e145811baaed775e57c08d57d81a34f5822d0c"}]}],"author":{"name":"matbu","email":"mat.bultel@gmail.com","date":"2022-07-15 11:54:48.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-08-01 09:00:51.000000000","tz":0},"subject":"Push 1.7.1 release for Validations Framework","message":"Push 1.7.1 release for Validations Framework\n\nPushed 1.7.1 release for validations framework:\n- validations-libs\n- validations-common\n\nChange-Id: Ie5d9eda9581471e754ffda4026adde7eade5226c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8048994bd536fd08b803e4636fff7e90a7de37ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8048994bd536fd08b803e4636fff7e90a7de37ec"}]},"branch":"refs/heads/master","description":"Edit commit message"},"536378904f2fa4801e74a7bb07ab2f47b474942c":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-08-01 09:00:55.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/94/849994/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/94/849994/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/849994/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/849994/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/94/849994/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/94/849994/8"}}},"commit":{"parents":[{"commit":"8bd6b930f53aa7d8b80718488fddeb4e216eb512","subject":"Merge \"[heat] Transition Pike to End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8bd6b930f53aa7d8b80718488fddeb4e216eb512"}]}],"author":{"name":"matbu","email":"mat.bultel@gmail.com","date":"2022-07-15 11:54:48.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-08-01 09:00:55.000000000","tz":0},"subject":"Push 1.7.1 release for Validations Framework","message":"Push 1.7.1 release for Validations Framework\n\nPushed 1.7.1 release for validations framework:\n- validations-libs\n- validations-common\n\nChange-Id: Ie5d9eda9581471e754ffda4026adde7eade5226c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/536378904f2fa4801e74a7bb07ab2f47b474942c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/536378904f2fa4801e74a7bb07ab2f47b474942c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[]}
