)]}'
{"id":"openstack%2Freleases~850753","triplet_id":"openstack%2Freleases~master~Iee5533b6e0a497b27774dedacb851789b9a6fb36","project":"openstack/releases","branch":"master","topic":"antelope-schedule-plan","attention_set":{},"removed_from_attention_set":{"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2022-09-02 15:20:42.000000000","reason":"Change was abandoned"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-09-02 15:20:24.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-09-02 15:20:42.000000000","reason":"Change was abandoned"},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2022-09-02 15:20:42.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iee5533b6e0a497b27774dedacb851789b9a6fb36","subject":"Proposed release schedule for Antelope (25w)","status":"ABANDONED","created":"2022-07-22 13:58:04.000000000","updated":"2022-09-02 15:20:42.000000000","total_comment_count":22,"unresolved_comment_count":5,"has_review_started":true,"meta_rev_id":"2a13fd3c0d8afea2c2b6ef9a284df92db1fbbc80","_number":850753,"virtual_id_number":850753,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-07-22 17:31:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":1,"date":"2022-08-08 21:59:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":1,"date":"2022-07-29 08:01:55.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2022-07-29 08:08:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-01 13:07:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"value":-1,"date":"2022-09-01 19:42:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":1,"date":"2022-08-03 12:03:56.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-22 14:05:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-22 14:12:20.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2022-07-29 07:41:16.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2022-07-29 08:01:55.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-07-29 08:08:31.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-08-01 12:10:30.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"CC"},{"updated":"2022-08-01 13:07:29.000000000","updated_by":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"reviewer":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"state":"REVIEWER"},{"updated":"2022-08-03 12:03:56.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2022-08-08 21:59:47.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2022-08-20 13:23:20.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"}],"messages":[{"id":"63e019a4a2a605949022b636f6537c9779d41e83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-22 13:58:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d41099e36eca76f7a14cd015c6dce45234c46c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 14:05:14.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ebc720f244c5400699ae42b1402dad2f\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7546aa1eab4b4cd8ac973a60e1b2a4d5 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4854d98029184abb998c3dcb6791254e : FAILURE in 5m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1e9249b250c04b0d8f06396d819f09f1 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":1},{"id":"88000acfa7e507b49ef19f664537e6a4ca0549f6","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-07-22 14:12:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af0d49ffb092f6b5dfc53a9a8a9882aaeebeab6f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-07-22 14:15:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"69c9df63a3f6cd906a232be3918bf3a3f9d503ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-22 14:55:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e99523f184417d1b89852013edca34b8a1861b44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 15:05:45.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/99530ebfaff644bc80f131cca7361f60\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/467290e1686d46b99b7375176a12480c : SUCCESS in 7m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6554c0b427a455e8a6bda9280f1982b : FAILURE in 5m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d359184469094d86a49fd46502044a74 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":2},{"id":"b47d1473c0e50e24a1a63874cfd3ae9aa3f92c51","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-07-22 15:24:15.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nSee inline comments for the source of the new doc build error.\n\nAlso, I checked with Allison and Jimmy, seems like the suggestion is to do 2023-03-29 instead (a week earlier) in order to have it over well before Passover and Easter, if you want to maximize marketing reach for the announcement. Doing it after the holidays you\u0027d run into people being focused on Kubecon EU and OCP EU (April 17-20).","accounts_in_message":[],"_revision_number":2},{"id":"6d464ce398803211720dc4367e2ed2745c32d335","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-22 15:54:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4be73306fd6b815e52ba3cda0dc5d9628e8f477b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 16:02:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0eadd9714114736976d92c3a4e3c3d9\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/011a2c25fd5d49439f0476c86908f850 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e81c618a69f04d62ade9f85bf5bd23eb : SUCCESS in 7m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cf5adcc0e8ae443a80cfda2c0d710a8b : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":3},{"id":"4c46dc3d9dc75e00ed7b7fc41f502015e1caf8f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-22 16:41:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bbf15d6f87c5b755ec937c317781f86573297873","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 16:49:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cf9556fc6124859a48a27132e0e8f88\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60708cfc016a42aab94faef367b90c6d : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa1ab91a37a44cdebda56cfc57996d73 : SUCCESS in 6m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4472a2e4c9b5480dbfb25c89d80a2d26 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":4},{"id":"38ab0c2b718b10ea04a60c9ad7354d9ce5d01169","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-22 17:15:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4c0bf8ec447c3b212681bd9845abc33ae04c5e69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 17:31:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/747e3dc9ad774efb91766f4a04aa5d28\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82aa76a11cd6480f8d9e043b755853b2 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/699fc2d5a58a46ea8e27e9d43a6bc196 : SUCCESS in 7m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0915bf3f826e4d44aaeb4b7279032e8d : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":5},{"id":"8ec76b71ae9169d493d8066da2b34f96b94a7cf5","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-07-29 07:41:16.000000000","message":"Patch Set 5: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"8dd770b3ebc5da95fcc41370d7af26bc0ed7f7c0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-07-29 08:01:55.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7f67715ef2c1bd369eecf812678622a3d28af9aa","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-07-29 08:08:31.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b5268c5a39fb401a7d0088ed1feb94087c1929fc","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-08-01 12:10:30.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ec9eeedaf2beacdd15ae259135605c61fd7601d7","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2022-08-01 13:07:29.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c46c599c7b762bb138f6e921c5bf8243932ef7b4","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-08-01 14:21:21.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7de5aab7fdf85c9f473f2caf2d3e6edb43ea49a3","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-08-02 09:11:18.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"50f180f3dd8752e0e66ea8685e23135906242abf","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-08-02 09:16:43.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ee439b94681de09b1acf281c0c50e62315ae168d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-08-03 12:03:56.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b3762ff03f17e758e71b253c2c05023fef80312f","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2022-08-08 21:59:47.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"3965656ac15445b49f97a33e82490f6a28d17070","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-08-11 09:08:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"49bd43c2b2d92db9b644d7a2f9985f8d6242e94e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-08-20 13:23:20.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"04ea0613400db0f01e218fcd38a3f97c1115b217","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-09-01 19:42:35.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"82cd5562d0c95e50b89f37241dcc30f952c63f54","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-02 15:20:24.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"2a13fd3c0d8afea2c2b6ef9a284df92db1fbbc80","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-02 15:20:42.000000000","message":"Abandoned\n\nsee reason in comments.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"528b4c351458962fd1d33bcdb827abc6ec86bb56","revisions":{"77acb36f500ef484f4a3c32d2af6876017dda13c":{"kind":"REWORK","_number":1,"created":"2022-07-22 13:58:04.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/53/850753/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/53/850753/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/850753/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/850753/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/850753/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/53/850753/1"}}},"commit":{"parents":[{"commit":"8f57ab1103e75054bd84ad743d185cd26e6505fa","subject":"Merge \"[manila] Release manila-tempest-plugin 1.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8f57ab1103e75054bd84ad743d185cd26e6505fa"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-07-22 13:48:31.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-07-22 13:51:46.000000000","tz":120},"subject":"WIP: Proposed release schedule for Antelope (26w)","message":"WIP: Proposed release schedule for Antelope (26w)\n\nThis is a 26 week schedule for the Antelope release\nPlanned release date is 05 April, 2023.\n\nChange-Id: Iee5533b6e0a497b27774dedacb851789b9a6fb36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/77acb36f500ef484f4a3c32d2af6876017dda13c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/77acb36f500ef484f4a3c32d2af6876017dda13c"}]},"branch":"refs/heads/master"},"524346e651590cf27472f49d7a23558a70b482c8":{"kind":"REWORK","_number":2,"created":"2022-07-22 14:55:16.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/53/850753/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/53/850753/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/850753/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/850753/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/850753/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/53/850753/2"}}},"commit":{"parents":[{"commit":"8f57ab1103e75054bd84ad743d185cd26e6505fa","subject":"Merge \"[manila] Release manila-tempest-plugin 1.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8f57ab1103e75054bd84ad743d185cd26e6505fa"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-07-22 13:48:31.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-07-22 14:54:48.000000000","tz":120},"subject":"WIP: Proposed release schedule for Antelope (26w)","message":"WIP: Proposed release schedule for Antelope (26w)\n\nThis is a 26 week schedule for the Antelope release\nPlanned release date is 05 April, 2023.\n\nChange-Id: Iee5533b6e0a497b27774dedacb851789b9a6fb36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/524346e651590cf27472f49d7a23558a70b482c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/524346e651590cf27472f49d7a23558a70b482c8"}]},"branch":"refs/heads/master"},"7415027789fa0653432efa291efdee72d9322a81":{"kind":"REWORK","_number":3,"created":"2022-07-22 15:54:22.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/53/850753/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/53/850753/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/850753/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/850753/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/850753/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/53/850753/3"}}},"commit":{"parents":[{"commit":"8f57ab1103e75054bd84ad743d185cd26e6505fa","subject":"Merge \"[manila] Release manila-tempest-plugin 1.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8f57ab1103e75054bd84ad743d185cd26e6505fa"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-07-22 13:48:31.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-07-22 15:53:59.000000000","tz":120},"subject":"WIP: Proposed release schedule for Antelope (26w)","message":"WIP: Proposed release schedule for Antelope (26w)\n\nThis is a 26 week schedule for the Antelope release\nPlanned release date is 05 April, 2023.\n\nChange-Id: Iee5533b6e0a497b27774dedacb851789b9a6fb36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7415027789fa0653432efa291efdee72d9322a81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7415027789fa0653432efa291efdee72d9322a81"}]},"branch":"refs/heads/master"},"efb3932133307d0e4a3bdebb6bae53c6e086a8ee":{"kind":"REWORK","_number":4,"created":"2022-07-22 16:41:58.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/53/850753/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/53/850753/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/850753/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/850753/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/850753/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/53/850753/4"}}},"commit":{"parents":[{"commit":"8f57ab1103e75054bd84ad743d185cd26e6505fa","subject":"Merge \"[manila] Release manila-tempest-plugin 1.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8f57ab1103e75054bd84ad743d185cd26e6505fa"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-07-22 13:48:31.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-07-22 16:41:13.000000000","tz":120},"subject":"Proposed release schedule for Antelope (25w)","message":"Proposed release schedule for Antelope (25w)\n\nThis is a 25 week schedule for the Antelope release\nPlanned release date is 29 March, 2023.\n\nChange-Id: Iee5533b6e0a497b27774dedacb851789b9a6fb36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/efb3932133307d0e4a3bdebb6bae53c6e086a8ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/efb3932133307d0e4a3bdebb6bae53c6e086a8ee"}]},"branch":"refs/heads/master"},"528b4c351458962fd1d33bcdb827abc6ec86bb56":{"kind":"REWORK","_number":5,"created":"2022-07-22 17:15:49.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/53/850753/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/53/850753/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/850753/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/850753/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/850753/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/53/850753/5"}}},"commit":{"parents":[{"commit":"8f57ab1103e75054bd84ad743d185cd26e6505fa","subject":"Merge \"[manila] Release manila-tempest-plugin 1.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8f57ab1103e75054bd84ad743d185cd26e6505fa"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-07-22 13:48:31.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-07-22 17:15:23.000000000","tz":120},"subject":"Proposed release schedule for Antelope (25w)","message":"Proposed release schedule for Antelope (25w)\n\nThis is a 25 week schedule for the Antelope release\nPlanned release date is 29 March, 2023.\n\nChange-Id: Iee5533b6e0a497b27774dedacb851789b9a6fb36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/528b4c351458962fd1d33bcdb827abc6ec86bb56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/528b4c351458962fd1d33bcdb827abc6ec86bb56"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
