)]}'
{"id":"openstack%2Freleases~850889","triplet_id":"openstack%2Freleases~master~Ida6310dd822e587ce9c424a228fdff5192ba0476","project":"openstack/releases","branch":"master","topic":"jsonschema-4-x","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-07-27 08:31:35.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-07-26 11:13:33.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},"30750":{"account":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"last_update":"2022-07-27 05:31:52.000000000","reason":"\u003cGERRIT_ACCOUNT_30750\u003e replied on the change","reason_account":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"}}},"hashtags":[],"change_id":"Ida6310dd822e587ce9c424a228fdff5192ba0476","subject":"Uncap jsonschema","status":"MERGED","created":"2022-07-25 13:42:57.000000000","updated":"2022-07-27 08:31:35.000000000","submitted":"2022-07-27 08:31:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850889-jsonschema-4-x","meta_rev_id":"60dafca04cdb1cc15a1494c036d05a6178934465","_number":850889,"virtual_id_number":850889,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-27 08:31:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},{"value":0,"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"recommended":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":1,"date":"2022-07-27 07:12:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2022-07-27 08:21:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2022-07-26 11:13:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-26 13:31:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},{"value":1,"date":"2022-07-27 05:31:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2022-07-27 08:21:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},{"value":0,"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},{"value":0,"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-25 13:51:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-26 06:20:26.000000000","updated_by":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"reviewer":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"state":"CC"},{"updated":"2022-07-26 07:28:57.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"state":"REVIEWER"},{"updated":"2022-07-26 09:12:38.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-07-27 05:31:52.000000000","updated_by":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"reviewer":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"state":"REVIEWER"},{"updated":"2022-07-27 07:12:57.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2022-07-27 08:21:32.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"439be6886ebe39e88badcf7e3c4936360e38bdd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-25 13:42:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b0c744d7575efc39a4754778f4e44e550231103","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 13:51:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7085520a82b4d1abba339777bac4617\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb8c5ac3373e486f882db7f06361a1f7 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf8a75ca00dc4d23bf3677edd91a038c : FAILURE in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad21b7aa86ab459f863dab6ee0309308 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cdfb3e21ca04af1813397bd3f0d5564 : SUCCESS in 7m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b1e23a635b904feb819f491d4e8f2209 : FAILURE in 4m 54s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0ccb10f9130e4703a6597409c0ee1486 : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b728958a81014c528314c6f9e9e41b16 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":1},{"id":"e49bbedcddaa3b4de5039cc0d2b5b7619e6594ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-25 15:59:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8f36f1a8f71e16268d7badce57d0fd6fbad084b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 16:09:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a78f9f158fa94d93af54ad363628c434\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92516127abc24109857fc447dc2135f2 : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c08963a1e9bc4e87b0a16efe018bb642 : FAILURE in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07143e924ea24eba96f402b2481a4644 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69c5e72ddbea468589c1fc7abb988db4 : SUCCESS in 7m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4b3f5aa648824029a6afaa43cf6b0432 : SUCCESS in 5m 32s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/dbc0de9865be4dbf8bd9f13f00222265 : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/85d413b871144e128cdff75abaf19b67 : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":2},{"id":"570ed41cf98a24a508dfff71d0887894903356e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-25 16:13:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a4bfa4321219a676172a38e27f8e844f1d143ff7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 16:23:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e77fb31b9934f67ac97a2ec3f58da67\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34cae8bc93e74b4d92deee021a86a8d0 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbbf5d3956a144dda37878723c30c5ca : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac20ebd9fb5f4782a93e48af5dc563f6 : SUCCESS in 5m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/efde120a54ec42bbba4e1f47714fbfb5 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48fb0abd2f2044d1b721b2634d17ff0b : SUCCESS in 7m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/124f70e041344e5dbe40bb171cba03a0 : SUCCESS in 5m 44s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/544c7ff9cc7f4426902ba24225385a2f : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76a785773bad4aea95ae06454bfc42bd : FAILURE in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/38eec8746ab84fd3bf6c540c7a75579a : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":3},{"id":"6e0ac0c33853728c3e4d5667c2f2a1721b809beb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-25 16:24:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6b0066781cfa06045ed377acb7c2143bfc5a5118","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 16:33:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f66230dede943d1ae8dfef99edd2eb5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bc2c909e1404ff1b56dde1347bc7e05 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4913259768e84c4d9ce0059f2efb8538 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c70f3dfe63b14f5b8e08f02bc6305ddb : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ac6cacb767445c1918d4e9076eb3a33 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e55d474417314a46ae66ba75825982fd : SUCCESS in 6m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1a9bb9aca41241f8a46128b3a10dad4b : SUCCESS in 5m 40s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2e810149033a4d0ea28bd2d52fb17994 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":4},{"id":"683a313bb26e2638895c4504650d501494b663da","author":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"date":"2022-07-26 06:20:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"49dab3420dc9d5db8b7d53e769aa7c2b212af29a","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-07-26 07:28:57.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"7b8a526e0c948aef3d3554aeb4e815a46ae410de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-26 10:24:48.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"01b8319bef628ab71a911ef57f02c2b8b307cecd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 10:37:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7afa5d3c39f49c380b5002eb1a117cf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f41ddcb3cfe492a8f7eec1476a1e432 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b2533fdca7e496ea1b9d6ecc900dd23 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63d1d9c4ada04dbaa7c5e44e1a9cbad2 : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/22292d66a85c40e5bbc5bb983bf20689 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48d184bed5fc4cfb8c104bd6fd1b8efd : SUCCESS in 7m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7d0673567337409abc6f2e16d2286a29 : SUCCESS in 5m 38s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f5fca8c1cebc449680d244f7ef73eef2 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":5},{"id":"f4d80bfa5848c9f64bbad58349fd4ea89e0b9239","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-26 10:38:37.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e370d4a1b4dcbcca08177772d91edca5a6cba051","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-07-26 11:13:33.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cb1bfd9a3bca047e2ca0052d5995f6c0c0956c04","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-07-26 13:31:34.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"7adaab0691146d35bc673feecad48bb2cc6fb84d","author":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"date":"2022-07-27 05:31:52.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"bb3d6ba00a98c768277929087fb3558473e40ca0","author":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"date":"2022-07-27 05:32:28.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5669f49e1a965b8ec3dff81576b78a6bc5746e08","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-07-27 07:12:57.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"d065b24f8cb19de6951be33579a433402e2572cb","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-07-27 08:21:32.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"1aec65742010e91ee3e2141a14a52d5fa2553316","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 08:21:58.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7f8c8d38dc18e383afd8993229e0103dbfb26fe0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 08:31:34.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/653e7ce690004413a6b80a53330c331f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c698fb5721a4927847928bf28c9ad49 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca5c8cd2ec8a4511b3232f7cca00c672 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45db28fdfafb44d48de3b95985686873 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1ee23f9343c434bbdc0c2359053fba5 : SUCCESS in 7m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a1d6001126474984b63c27d1d7411641 : SUCCESS in 5m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a8f75fbd33d48039bd0f2ca86c416a2 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":5},{"id":"60dafca04cdb1cc15a1494c036d05a6178934465","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 08:31:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e83bb1f69a4612941c0aec519452f3e6cf7c5cd4","revisions":{"b8a5494b54ad2d7dec6a50e15ee6eba0c7154dbc":{"kind":"REWORK","_number":1,"created":"2022-07-25 13:42:57.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/89/850889/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/89/850889/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/850889/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/850889/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/850889/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/89/850889/1"}}},"commit":{"parents":[{"commit":"5d54e9738311d795fa87802c910556e06cfa7912","subject":"Merge \"Update _CLOSED_SERIES set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5d54e9738311d795fa87802c910556e06cfa7912"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-07-25 13:38:25.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-07-25 13:38:25.000000000","tz":120},"subject":"Updating jsonschema constraint","message":"Updating jsonschema constraint\n\nWith the latest upper constraints updating generated patch [1] the\nopenstack-tox-docs job started to fail, due to conflict:\n\nERROR: Cannot install jsonschema\u003c4.0.0 and \u003e\u003d3.2.0 because these\npackage versions have conflicting dependencies.\n\nThe conflict is caused by:\n    The user requested jsonschema\u003c4.0.0 and \u003e\u003d3.2.0\n    The user requested (constraint) jsonschema\u003d\u003d\u003d4.7.2\n\nThis is because we have jsonschema\u003c4.0.0 in our requirements.txt.\nBy updating the constraint to be \u003e4.0.0 the docs job is passing.\n\n[1] I89576353371aa87d181610eb6242cbe961487af6\n\nChange-Id: Ida6310dd822e587ce9c424a228fdff5192ba0476\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b8a5494b54ad2d7dec6a50e15ee6eba0c7154dbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b8a5494b54ad2d7dec6a50e15ee6eba0c7154dbc"}]},"branch":"refs/heads/master"},"ee460bd5b0d762e806e4901cf10b6a770b2dbb88":{"kind":"REWORK","_number":2,"created":"2022-07-25 15:59:52.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/89/850889/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/89/850889/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/850889/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/850889/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/850889/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/89/850889/2"}}},"commit":{"parents":[{"commit":"5d54e9738311d795fa87802c910556e06cfa7912","subject":"Merge \"Update _CLOSED_SERIES set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5d54e9738311d795fa87802c910556e06cfa7912"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-07-25 13:38:25.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-07-25 15:54:56.000000000","tz":120},"subject":"Updating jsonschema constraint","message":"Updating jsonschema constraint\n\nWith the latest upper constraints updating generated patch [1] the\nopenstack-tox-docs job started to fail, due to conflict:\n\nERROR: Cannot install jsonschema\u003c4.0.0 and \u003e\u003d3.2.0 because these\npackage versions have conflicting dependencies.\n\nThe conflict is caused by:\n    The user requested jsonschema\u003c4.0.0 and \u003e\u003d3.2.0\n    The user requested (constraint) jsonschema\u003d\u003d\u003d4.7.2\n\nThis is because we have jsonschema\u003c4.0.0 in our requirements.txt.\nBy updating the constraint to be \u003e4.0.0 the docs job is passing.\n\nDate validation needed to be updated as well, because a deprecated\nusage of validator extending method was used in the code, which was\nremoved [2] from jsonschema with the 4.0.0 release. This patch replaces\nthat with adding a TYPE_CHECKER instead.\n\n[1] I89576353371aa87d181610eb6242cbe961487af6\n[2] https://github.com/python-jsonschema/jsonschema/commit/d9f63840609176f0b7328f2088c3b2a4016526cd\n\nChange-Id: Ida6310dd822e587ce9c424a228fdff5192ba0476\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ee460bd5b0d762e806e4901cf10b6a770b2dbb88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ee460bd5b0d762e806e4901cf10b6a770b2dbb88"}]},"branch":"refs/heads/master"},"7317d111fbe93d8b0d936212d19fe2f0fad1ee3f":{"kind":"REWORK","_number":3,"created":"2022-07-25 16:13:53.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/89/850889/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/89/850889/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/850889/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/850889/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/850889/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/89/850889/3"}}},"commit":{"parents":[{"commit":"5d54e9738311d795fa87802c910556e06cfa7912","subject":"Merge \"Update _CLOSED_SERIES set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5d54e9738311d795fa87802c910556e06cfa7912"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-07-25 13:38:25.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-07-25 16:04:05.000000000","tz":120},"subject":"Updating jsonschema constraint","message":"Updating jsonschema constraint\n\nWith the latest upper constraints updating generated patch [1] the\nopenstack-tox-docs job started to fail, due to conflict:\n\nERROR: Cannot install jsonschema\u003c4.0.0 and \u003e\u003d3.2.0 because these\npackage versions have conflicting dependencies.\n\nThe conflict is caused by:\n    The user requested jsonschema\u003c4.0.0 and \u003e\u003d3.2.0\n    The user requested (constraint) jsonschema\u003d\u003d\u003d4.7.2\n\nThis is because we have jsonschema\u003c4.0.0 in our requirements.txt.\nBy updating the constraint to be \u003e4.0.0 the docs job is passing.\n\nDate validation needed to be updated as well, because a deprecated\nusage of validator extending method was used in the code, which was\nremoved [2] from jsonschema with the 4.0.0 release. This patch replaces\nthat with adding a TYPE_CHECKER instead.\n\nWith the new jsonschema constraint openstack-tox-py36 job does not pass\nanymore, but py36 is dropped from zed anyway. That job is pulled in by\nan old job template, which needs to be replaced to the new zed job\ntemplates, too.\n\n[1] I89576353371aa87d181610eb6242cbe961487af6\n[2] https://github.com/python-jsonschema/jsonschema/commit/d9f63840609176f0b7328f2088c3b2a4016526cd\n\nChange-Id: Ida6310dd822e587ce9c424a228fdff5192ba0476\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7317d111fbe93d8b0d936212d19fe2f0fad1ee3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7317d111fbe93d8b0d936212d19fe2f0fad1ee3f"}]},"branch":"refs/heads/master"},"73bff4cfee71c35fd92fbff97f45ddf07feef41c":{"kind":"REWORK","_number":4,"created":"2022-07-25 16:24:28.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/89/850889/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/89/850889/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/850889/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/850889/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/850889/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/89/850889/4"}}},"commit":{"parents":[{"commit":"5d54e9738311d795fa87802c910556e06cfa7912","subject":"Merge \"Update _CLOSED_SERIES set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5d54e9738311d795fa87802c910556e06cfa7912"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-07-25 13:38:25.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-07-25 16:22:32.000000000","tz":120},"subject":"Updating jsonschema constraint","message":"Updating jsonschema constraint\n\nWith the latest upper constraints updating generated patch [1] the\nopenstack-tox-docs job started to fail, due to conflict:\n\nERROR: Cannot install jsonschema\u003c4.0.0 and \u003e\u003d3.2.0 because these\npackage versions have conflicting dependencies.\n\nThe conflict is caused by:\n    The user requested jsonschema\u003c4.0.0 and \u003e\u003d3.2.0\n    The user requested (constraint) jsonschema\u003d\u003d\u003d4.7.2\n\nThis is because we have jsonschema\u003c4.0.0 in our requirements.txt.\nBy updating the constraint to be \u003e4.0.0 the docs job is passing.\n\nDate validation needed to be updated as well, because a deprecated\nusage of validator extending method was used in the code, which was\nremoved [2] from jsonschema with the 4.0.0 release. This patch replaces\nthat with adding a TYPE_CHECKER instead.\n\nWith the new jsonschema constraint openstack-tox-py36 job does not pass\nanymore, but py36 is dropped from zed anyway. That job is pulled in by\nan old job template, which needs to be replaced to the new zed job\ntemplates, too.\n\n[1] I89576353371aa87d181610eb6242cbe961487af6\n[2] https://github.com/python-jsonschema/jsonschema/commit/d9f63840609176f0b7328f2088c3b2a4016526cd\n\nChange-Id: Ida6310dd822e587ce9c424a228fdff5192ba0476\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/73bff4cfee71c35fd92fbff97f45ddf07feef41c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/73bff4cfee71c35fd92fbff97f45ddf07feef41c"}]},"branch":"refs/heads/master"},"e83bb1f69a4612941c0aec519452f3e6cf7c5cd4":{"kind":"REWORK","_number":5,"created":"2022-07-26 10:24:48.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/89/850889/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/89/850889/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/850889/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/850889/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/850889/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/89/850889/5"}}},"commit":{"parents":[{"commit":"5d54e9738311d795fa87802c910556e06cfa7912","subject":"Merge \"Update _CLOSED_SERIES set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5d54e9738311d795fa87802c910556e06cfa7912"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-07-25 13:38:25.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-07-26 10:24:11.000000000","tz":120},"subject":"Uncap jsonschema","message":"Uncap jsonschema\n\nWith the latest upper constraints updating generated patch [1] the\nopenstack-tox-docs job started to fail, due to conflict:\n\nERROR: Cannot install jsonschema\u003c4.0.0 and \u003e\u003d3.2.0 because these\npackage versions have conflicting dependencies.\n\nThe conflict is caused by:\n    The user requested jsonschema\u003c4.0.0 and \u003e\u003d3.2.0\n    The user requested (constraint) jsonschema\u003d\u003d\u003d4.7.2\n\nThis is because we have jsonschema\u003c4.0.0 in our requirements.txt.\nBy uncapping the constraint the docs job is passing.\n\nDate validation needed to be updated as well, because a deprecated\nusage of validator extending method was used in the code, which was\nremoved [2] from jsonschema with the 4.0.0 release. This patch replaces\nthat with adding a TYPE_CHECKER instead.\n\nWith the new jsonschema constraint openstack-tox-py36 job does not pass\nanymore, but py36 is dropped from zed anyway. That job is pulled in by\nan old job template, which needs to be replaced to the new zed job\ntemplates, too.\n\n[1] I89576353371aa87d181610eb6242cbe961487af6\n[2] https://github.com/python-jsonschema/jsonschema/commit/d9f63840609176f0b7328f2088c3b2a4016526cd\n\nCloses-Bug: #1982757\nChange-Id: Ida6310dd822e587ce9c424a228fdff5192ba0476\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e83bb1f69a4612941c0aec519452f3e6cf7c5cd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e83bb1f69a4612941c0aec519452f3e6cf7c5cd4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[]}
