)]}'
{"id":"openstack%2Freleases~850891","triplet_id":"openstack%2Freleases~master~Icc22e9e79b2f5a77889855e5a7cc1d21e9c5ee0a","project":"openstack/releases","branch":"master","topic":"deps/jsonschema","attention_set":{},"removed_from_attention_set":{"32926":{"account":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"last_update":"2022-07-26 06:29:13.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Icc22e9e79b2f5a77889855e5a7cc1d21e9c5ee0a","subject":"Uncapping jsonschema","status":"ABANDONED","created":"2022-07-25 13:49:51.000000000","updated":"2022-07-26 06:29:13.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3a8451fed79b98d64ebc96e2165b6489d6e8d7b3","_number":850891,"virtual_id_number":850891,"owner":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-07-25 14:01:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":-1,"date":"2022-07-25 16:04:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-25 14:01:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-25 16:04:54.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"3941b67e46bde507bda3bb01f6574176bdf61d78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-07-25 13:49:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ebe5cf35032e2ff66ef5a484828613e868b7b5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 14:01:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/baced37ce0394a478cbf51170f3ad851\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf3a85d9d1de42d1b67cf6cf73212baf : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae103c1affba469c90c89d4ff1a460de : SUCCESS in 7m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f62aaf9a46244a87a2bacfef550768d1 : SUCCESS in 7m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cc02e7c8850456693e89cd6b1eaaa38 : SUCCESS in 9m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8da22521846949e0a63812498fd8c502 : FAILURE in 7m 05s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/546c79f6384e4217954416df1b63daa8 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d375d33b24194983896de8a241e73e9e : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":1},{"id":"f1f2a1a6f11cbadac16947741d51686c058261fc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-07-25 16:04:54.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3a8451fed79b98d64ebc96e2165b6489d6e8d7b3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-07-26 06:29:13.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8fce4e272d13d3c7f4fc69f8f27f184e76ee560f","revisions":{"8fce4e272d13d3c7f4fc69f8f27f184e76ee560f":{"kind":"REWORK","_number":1,"created":"2022-07-25 13:49:51.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/91/850891/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/91/850891/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/850891/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/850891/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/850891/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/91/850891/1"}}},"commit":{"parents":[{"commit":"5d54e9738311d795fa87802c910556e06cfa7912","subject":"Merge \"Update _CLOSED_SERIES set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5d54e9738311d795fa87802c910556e06cfa7912"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-07-25 13:49:20.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-07-25 13:49:21.000000000","tz":120},"subject":"Uncapping jsonschema","message":"Uncapping jsonschema\n\nCloses-Bug: #1982757\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: Icc22e9e79b2f5a77889855e5a7cc1d21e9c5ee0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8fce4e272d13d3c7f4fc69f8f27f184e76ee560f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8fce4e272d13d3c7f4fc69f8f27f184e76ee560f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
