)]}'
{"id":"openstack%2Freleases~851165","triplet_id":"openstack%2Freleases~master~I6ed0bbd112625cdc9aafbbdf36e2d4304fdaaae2","project":"openstack/releases","branch":"master","topic":"release/14.2.2","hashtags":[],"change_id":"I6ed0bbd112625cdc9aafbbdf36e2d4304fdaaae2","subject":"Release tripleo-validations for stable/wallaby","status":"ABANDONED","created":"2022-07-27 07:32:48.000000000","updated":"2022-07-27 09:21:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"9872def82335a6e996e2e79114a3f25e27c25764","_number":851165,"virtual_id_number":851165,"owner":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-07-27 07:55:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2022-07-27 07:34:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-27 07:43:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b0cc2bd2c6b3660cef1d23d98eed00f951b2e9c8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-07-27 07:32:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c755f8145c4899ea10b7204dda6fb699f88ce1d0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-07-27 07:35:01.000000000","message":"Topic release/14.3.0 removed","accounts_in_message":[],"_revision_number":1},{"id":"47b397897a37daa22ade6a1553bdfd2c71515ef3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-07-27 07:35:11.000000000","message":"Topic set to release/14.2.2","accounts_in_message":[],"_revision_number":1},{"id":"eee75d513d591539c2b08efbf3f50505b88c58d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 07:43:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce8b74407df04755a45078ad2f54c257\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90e305f746e14ab187cabb87d44a55a4 : FAILURE in 3m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9ec03b3128ef43ed9738681a9d62a81a : SUCCESS in 5m 37s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a41a903367b647a08c976a8e9dab99f8 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"4aa0683bdc99155ebbd8119b81ab302e8d8312fe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-07-27 07:46:49.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"816781bca3efea4ba9eaa1ce05daa776ed074f76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 07:55:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74daa6d611b6464fae8a8d0050bb8002\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbf49f851b4c46edab4637c46552b214 : SUCCESS in 7m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/23e8596291a34e9bba039cb4b52fbd49 : SUCCESS in 6m 13s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/28e557a7df9f4104989a06c46dad2ec5 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":2},{"id":"9872def82335a6e996e2e79114a3f25e27c25764","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-07-27 09:21:54.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7baee29ef889b53157b18f57dd834d8664d6d376","revisions":{"8e15bf031057525b4ee2c568c63a5db8bd2926ff":{"kind":"REWORK","_number":1,"created":"2022-07-27 07:32:48.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/65/851165/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/65/851165/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/851165/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/851165/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/851165/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/65/851165/1"}}},"commit":{"parents":[{"commit":"5d54e9738311d795fa87802c910556e06cfa7912","subject":"Merge \"Update _CLOSED_SERIES set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5d54e9738311d795fa87802c910556e06cfa7912"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-07-27 06:58:35.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-07-27 07:31:56.000000000","tz":120},"subject":"Release tripleo-validations for stable/wallaby","message":"Release tripleo-validations for stable/wallaby\n\nThis release include fixes for pacemaker and nova validations,\nas well as minor adjustments to documentation.\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I6ed0bbd112625cdc9aafbbdf36e2d4304fdaaae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8e15bf031057525b4ee2c568c63a5db8bd2926ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8e15bf031057525b4ee2c568c63a5db8bd2926ff"}]},"branch":"refs/heads/master"},"7baee29ef889b53157b18f57dd834d8664d6d376":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-07-27 07:46:49.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/65/851165/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/65/851165/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/851165/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/851165/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/851165/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/65/851165/2"}}},"commit":{"parents":[{"commit":"5d54e9738311d795fa87802c910556e06cfa7912","subject":"Merge \"Update _CLOSED_SERIES set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5d54e9738311d795fa87802c910556e06cfa7912"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-07-27 06:58:35.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-07-27 07:46:09.000000000","tz":120},"subject":"Release tripleo-validations for stable/wallaby","message":"Release tripleo-validations for stable/wallaby\n\nThis release include fixes for pacemaker and nova validations,\nas well as minor adjustments to documentation.\n\nDepends-On: https://review.opendev.org/c/openstack/releases/+/850889\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I6ed0bbd112625cdc9aafbbdf36e2d4304fdaaae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7baee29ef889b53157b18f57dd834d8664d6d376"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7baee29ef889b53157b18f57dd834d8664d6d376"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
