)]}'
{"id":"openstack%2Freleases~851390","triplet_id":"openstack%2Freleases~master~I98e9a116f3f520d2083e0fb164c5cd39fde3188e","project":"openstack/releases","branch":"master","topic":"os-traits","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2022-07-28 11:25:31.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2022-07-28 15:09:28.000000000","reason":"Change was submitted"},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2022-07-28 14:30:49.000000000","reason":"\u003cGERRIT_ACCOUNT_7166\u003e replied on the change","reason_account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}}},"hashtags":[],"change_id":"I98e9a116f3f520d2083e0fb164c5cd39fde3188e","subject":"Release os-traits 2.8.0","status":"MERGED","created":"2022-07-28 11:18:38.000000000","updated":"2022-07-28 15:09:28.000000000","submitted":"2022-07-28 15:09:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851390-os-traits","meta_rev_id":"468f6eadfcc59e864cb61ee624f35e5a2dfa15b4","_number":851390,"virtual_id_number":851390,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-28 15:09:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"recommended":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2022-07-28 14:39:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2022-07-28 14:47:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2022-07-28 15:00:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-28 14:30:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2022-07-28 15:00:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-07-28 14:32:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-28 11:19:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-28 11:20:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2022-07-28 11:20:09.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2022-07-28 11:27:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-28 14:47:16.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2022-07-28 15:00:53.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"b94db377066bbcfff3a6ae3aa8cc917f7e515f86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-07-28 11:18:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3cc8e5b011c587c0759632fac840fb747e25ad3","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-07-28 11:19:42.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e015ac18f8c4a88bf00acd8e0195be7c006d07c1","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-07-28 11:25:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5d72df35b3b7fabaeb8b7a9877bd2c9b3455b716","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 11:27:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13a189084e654a59a175cbaa84133c7c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc94091e9e1449ed824c18b9d123cd91 : SUCCESS in 7m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e969c7faf3e14752b34cdc3cee17c110 : SUCCESS in 5m 00s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6d49365fab1f4e2f8d89e541a84536d2 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":1},{"id":"66140a6ff40ee19246495bf794258cbff9b3bb72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-07-28 14:29:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"25267a9889fae08b4c4b000ecc5aae7942ccc69b","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-07-28 14:30:49.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"cac329707f00c8ae418297248ad8db187c3f99a8","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 14:32:07.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"368694f8c8a58b27aee38a2e2feedfbf237ed1b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 14:38:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/407bcc3c1bc24555a5c4f66a3dc0aee7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f13f7b5f37e4870a1426457c1758f67 : SUCCESS in 6m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c06e528d53d2451d9df12534418ffc4a : SUCCESS in 4m 46s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9207905568434771aad56fcef65357d1 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":2},{"id":"1dfdcdbf27591ca8c5fe59b3c8945b8211a6996e","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-07-28 14:39:19.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6a8ab7c890685968f4da4bbc394c6ec9413729e5","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-07-28 14:47:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"14bd100b75bd5071421205c0d44e530e0f7e461d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-28 15:00:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"28833d98358640010ab057e3fbee70b4325824ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 15:01:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"90b11e50042a079422e1044a05f369376ef208a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 15:09:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/297b90ee061e4afaa5df351d29359ddf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abd8703d0bb54025bb2879fd2674fa4e : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdb5f6b76c8d45b0b75488ae343f9e36 : SUCCESS in 7m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5734fdda54fb4f7e8ba743df2185e389 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":2},{"id":"468f6eadfcc59e864cb61ee624f35e5a2dfa15b4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 15:09:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"072e85f2437f866bab9869c293bca34987ca5280","revisions":{"9c9caea496f8912eef7d1fa56201c29c0904c14e":{"kind":"REWORK","_number":1,"created":"2022-07-28 11:18:38.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/90/851390/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/90/851390/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/851390/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/851390/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/851390/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/90/851390/1"}}},"commit":{"parents":[{"commit":"b6ca6fef1545ede5c6ab25a68a052fd9f1a990e4","subject":"Merge \"EOL Kolla Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b6ca6fef1545ede5c6ab25a68a052fd9f1a990e4"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2022-07-28 11:17:02.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2022-07-28 11:18:30.000000000","tz":120},"subject":"Release os-traits 2.8.0","message":"Release os-traits 2.8.0\n\nThis release includes the following commits:\n$ git log --oneline --no-merges 2.7.0..HEAD\n62bb868 Add COMPUTE_MANAGED_PCI_DEVICE trait\na097c3f Add \u0027COMPUTE_STORAGE_VIRTIO_FS\u0027, \u0027COMPUTE_MEM_BACKING_FILE\u0027\nf64d50e Add nova and cyborg owner traits for resources\n87b1165 Remove unnecessary unicode prefixes\n1757c64 Update CI to use unversioned jobs template\n\nChange-Id: I98e9a116f3f520d2083e0fb164c5cd39fde3188e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9c9caea496f8912eef7d1fa56201c29c0904c14e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9c9caea496f8912eef7d1fa56201c29c0904c14e"}]},"branch":"refs/heads/master"},"072e85f2437f866bab9869c293bca34987ca5280":{"kind":"REWORK","_number":2,"created":"2022-07-28 14:29:37.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/90/851390/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/90/851390/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/851390/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/851390/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/90/851390/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/90/851390/2"}}},"commit":{"parents":[{"commit":"b6ca6fef1545ede5c6ab25a68a052fd9f1a990e4","subject":"Merge \"EOL Kolla Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b6ca6fef1545ede5c6ab25a68a052fd9f1a990e4"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2022-07-28 14:29:05.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2022-07-28 14:29:05.000000000","tz":120},"subject":"Release os-traits 2.8.0","message":"Release os-traits 2.8.0\n\nThis release includes the following commits:\n$ git log --oneline --no-merges 2.7.0..HEAD\n74988cd Add traits for vIOMMU\n62bb868 Add COMPUTE_MANAGED_PCI_DEVICE trait\na097c3f Add \u0027COMPUTE_STORAGE_VIRTIO_FS\u0027, \u0027COMPUTE_MEM_BACKING_FILE\u0027\nf64d50e Add nova and cyborg owner traits for resources\n87b1165 Remove unnecessary unicode prefixes\n1757c64 Update CI to use unversioned jobs template\n\nChange-Id: I98e9a116f3f520d2083e0fb164c5cd39fde3188e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/072e85f2437f866bab9869c293bca34987ca5280"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/072e85f2437f866bab9869c293bca34987ca5280"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[]}
