)]}'
{"id":"openstack%2Freleases~853156","triplet_id":"openstack%2Freleases~master~Ic71657f4d286361952f09cf9fbe84776da1c0c67","project":"openstack/releases","branch":"master","topic":"venus-deliverables-1st-release","attention_set":{},"removed_from_attention_set":{"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2022-08-18 09:16:07.000000000","reason":"\u003cGERRIT_ACCOUNT_308\u003e replied on the change","reason_account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-08-26 11:38:20.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2022-08-16 10:31:31.000000000","reason":"\u003cGERRIT_ACCOUNT_26458\u003e replied on the change","reason_account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}},"24061":{"account":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"last_update":"2022-08-25 00:49:19.000000000","reason":"\u003cGERRIT_ACCOUNT_24061\u003e replied on the change","reason_account":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"}}},"hashtags":[],"change_id":"Ic71657f4d286361952f09cf9fbe84776da1c0c67","subject":"Release venus-dashboard 1.0.0 beta for Zed","status":"MERGED","created":"2022-08-15 10:03:28.000000000","updated":"2022-08-26 12:03:08.000000000","submitted":"2022-08-26 12:03:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"853156-venus-deliverables-1st-release","meta_rev_id":"f4ebb911c99ef95e3fd8a56d90bfd719e514818c","_number":853156,"virtual_id_number":853156,"owner":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-26 12:03:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2022-08-18 09:16:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2022-08-24 09:26:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":2,"date":"2022-08-26 11:38:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2022-08-16 10:31:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-25 00:49:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":1,"date":"2022-08-26 11:38:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-08-25 00:50:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-15 10:04:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-15 14:50:38.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2022-08-16 01:01:01.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2022-08-16 01:01:01.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2022-08-24 09:26:08.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"}],"messages":[{"id":"a0d4d77d4467c19823680412c12611bdf309d63b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"date":"2022-08-15 10:03:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0e64819bb940597501a20f8e8dd8dc9fdf089d2","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 10:04:20.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"397870ff7a17f141dbc84330c1d9ea2030532334","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 10:16:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/811ead048df74618a6ae838465dd3099\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae9afc85eda14944be607e8db2d493c5 : SUCCESS in 7m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/23116e1c4f054e458dc475e8e52319ef : FAILURE in 6m 18s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b7c42ff13cfa465cb66f72b7c21c7c3e : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":1},{"id":"b5a41904fab0cf534311ed2c033482df36c95404","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-08-15 14:50:38.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"41c2ab58f7fab57abd382d2f90f2ac210f12b78d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"date":"2022-08-16 07:35:42.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"85505f38716412fc030ea08173cf2d94b288c8fb","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 07:36:40.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"82af5802bfc398582ff7b33ac17b962c2e37ae10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 07:47:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2b4334b12e14ebebb99eb49d8e4e66a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2982272f552441868058f809b67edd07 : SUCCESS in 6m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c7706615b246402698caa363732935f2 : SUCCESS in 5m 56s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/68e6ff8c1afe4052b7ebb396cd754331 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":2},{"id":"8242e5000258ef8f15382bde827e422bb7d98534","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-08-16 08:12:34.000000000","message":"Topic set to venus-deliverables-1st-release","accounts_in_message":[],"_revision_number":2},{"id":"22bf8be3c9e78aeb680ec36ad75218fe119880f7","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-08-16 09:08:50.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b97b52793bc83e34c463b582c1a9d0d014cd0e69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-08-16 09:49:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e1e88edfc27b13bb50d47cc7090e14f28d2aa179","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-08-16 09:51:06.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9e85e919a0ef60d32658fdc2966794a75234126f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 10:00:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/465bb94e106149d6b5aa128680fba7ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b8e66ba24514f2dbca134ff73de3a65 : SUCCESS in 9m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/92aec91cbbf24397a5856dd4ca576b4a : SUCCESS in 5m 48s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/46129b1e7d6342efbba7418313510dc2 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":3},{"id":"f8f2d2284592212cd7d59ee231f1a4c918d4683e","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-08-16 10:31:31.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"425d2c3708eda5aefa4d98e5ab569ef32b4df509","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-08-18 09:16:07.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"89905fac087321641c041351b7da30d5331e488b","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-08-24 09:26:08.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"9d471d22b99b4e127a148a1389c3ee5e62c9bec5","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-08-24 13:39:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"95a1374364986f1d945516f88a905c192c28b637","author":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"date":"2022-08-25 00:49:19.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"24246499cf646313ae2a6ecfbe4bcf500745e736","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 00:50:04.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"13fd194dd67aee4c38a660908b2090f1246e4f01","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-08-26 11:38:20.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d6bbd488a41efe786195964434acc6bdfd23e5e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 11:38:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0b5bcb5e453fbaf08f08f2fa3e15f4dfc0bb3449","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 12:03:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f33d60e11da4bc0ba098c26c829243b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ccdf0aea92e42a4a564cf78e10679ed : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94335450801543aea003850b7361df98 : SUCCESS in 8m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/73096f7909624ceabdee7b41c475ed9a : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":3},{"id":"f4ebb911c99ef95e3fd8a56d90bfd719e514818c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 12:03:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"81d6905da57d299855e4470a43e65c4a80f8fd12","revisions":{"082ba1c4328bb06ee14a7d41549c6368f184083b":{"kind":"REWORK","_number":1,"created":"2022-08-15 10:03:28.000000000","uploader":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"ref":"refs/changes/56/853156/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/56/853156/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/853156/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/853156/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/853156/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/56/853156/1"}}},"commit":{"parents":[{"commit":"4502c0728987b7da8d0c1d2c4a538d8d78e59f6f","subject":"Merge \"cliff 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4502c0728987b7da8d0c1d2c4a538d8d78e59f6f"}]}],"author":{"name":"pangliye","email":"pangliye@inspur.com","date":"2022-08-15 10:00:38.000000000","tz":480},"committer":{"name":"pangliye","email":"pangliye@inspur.com","date":"2022-08-15 10:03:20.000000000","tz":480},"subject":"Zed-3 Release for venus-dashboard","message":"Zed-3 Release for venus-dashboard\n\nThis is the Zed milestone-3 release for venus-dashboard,which is the first\nversion of venus-dashboard.\nThis venus-dashboard developed the display of log management.\n\nChange-Id: Ic71657f4d286361952f09cf9fbe84776da1c0c67\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/082ba1c4328bb06ee14a7d41549c6368f184083b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/082ba1c4328bb06ee14a7d41549c6368f184083b"}]},"branch":"refs/heads/master"},"bb77afcdd95ce9a41a903d1a9cf61d248eff1541":{"kind":"REWORK","_number":2,"created":"2022-08-16 07:35:42.000000000","uploader":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"ref":"refs/changes/56/853156/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/56/853156/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/853156/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/853156/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/853156/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/56/853156/2"}}},"commit":{"parents":[{"commit":"4502c0728987b7da8d0c1d2c4a538d8d78e59f6f","subject":"Merge \"cliff 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4502c0728987b7da8d0c1d2c4a538d8d78e59f6f"}]}],"author":{"name":"pangliye","email":"pangliye@inspur.com","date":"2022-08-15 10:00:38.000000000","tz":480},"committer":{"name":"pangliye","email":"pangliye@inspur.com","date":"2022-08-16 07:26:44.000000000","tz":0},"subject":"Release venus-dashboard 1.0.0 for Zed","message":"Release venus-dashboard 1.0.0 for Zed\n\nThis is the Zed milestone-3 release for venus-dashboard,which is the first\nversion of venus-dashboard.\nThis venus-dashboard developed the display of log management.\n\nChange-Id: Ic71657f4d286361952f09cf9fbe84776da1c0c67\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bb77afcdd95ce9a41a903d1a9cf61d248eff1541"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bb77afcdd95ce9a41a903d1a9cf61d248eff1541"}]},"branch":"refs/heads/master"},"81d6905da57d299855e4470a43e65c4a80f8fd12":{"kind":"REWORK","_number":3,"created":"2022-08-16 09:49:52.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/56/853156/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/56/853156/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/853156/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/853156/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/853156/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/56/853156/3"}}},"commit":{"parents":[{"commit":"4502c0728987b7da8d0c1d2c4a538d8d78e59f6f","subject":"Merge \"cliff 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4502c0728987b7da8d0c1d2c4a538d8d78e59f6f"}]}],"author":{"name":"pangliye","email":"pangliye@inspur.com","date":"2022-08-15 10:00:38.000000000","tz":480},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-08-16 09:48:24.000000000","tz":120},"subject":"Release venus-dashboard 1.0.0 beta for Zed","message":"Release venus-dashboard 1.0.0 beta for Zed\n\nThis is the first beta release for Zed for venus-dashboard. This\nvenus-dashboard developed the display of log management.\n\nChange-Id: Ic71657f4d286361952f09cf9fbe84776da1c0c67\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/81d6905da57d299855e4470a43e65c4a80f8fd12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/81d6905da57d299855e4470a43e65c4a80f8fd12"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[]}
