)]}'
{"id":"openstack%2Freleases~853158","triplet_id":"openstack%2Freleases~master~I62bb3606bed4aa3bf20f4cdb39a679bab89d7ce6","project":"openstack/releases","branch":"master","topic":"venus-deliverables-1st-release","attention_set":{},"removed_from_attention_set":{"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2022-08-16 09:09:20.000000000","reason":"\u003cGERRIT_ACCOUNT_308\u003e replied on the change","reason_account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-08-26 10:22:00.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"24061":{"account":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"last_update":"2022-08-25 00:49:37.000000000","reason":"\u003cGERRIT_ACCOUNT_24061\u003e replied on the change","reason_account":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"}}},"hashtags":[],"change_id":"I62bb3606bed4aa3bf20f4cdb39a679bab89d7ce6","subject":"Release python-venusclient 1.0.0 for Zed","status":"MERGED","created":"2022-08-15 10:17:00.000000000","updated":"2022-08-26 10:51:11.000000000","submitted":"2022-08-26 10:29:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"853158-venus-deliverables-1st-release","meta_rev_id":"995baa3e24baa1170ec9a9d32bb6a21cf83662fc","_number":853158,"virtual_id_number":853158,"owner":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-26 10:29:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2022-08-16 09:09:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2022-08-24 09:26:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":2,"date":"2022-08-26 10:22:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2022-08-16 08:12:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-25 00:49:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":1,"date":"2022-08-26 10:22:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-08-25 00:50:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-15 10:17:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-15 14:15:54.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2022-08-16 00:51:15.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2022-08-16 00:58:59.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2022-08-24 09:26:43.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"}],"messages":[{"id":"681389c5981cd0c56843117c766d73d243f6fafc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"date":"2022-08-15 10:17:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a312d08606ce6e4863e112d48b6cb1b461e1e53","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 10:17:56.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"6d3b65be26a1d7e536441179ead1783c9a3afcda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 10:30:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f748a01cc49d47fdaac6b36cb9ddadc5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/913087d9b69e4fffadf1e202a5d92493 : SUCCESS in 6m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/50135fbadc534c999bbef4612c975084 : SUCCESS in 6m 08s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/80794b5b20a840eab9ca6115fee29b98 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"25e3e0a66abc86851ae30ca3cffa4b8422aba221","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-08-15 14:15:54.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f76e35aa1ca9a441ec644944556a1cf957d2a50b","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-08-16 00:51:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"659c9246849024201635b95f674c269bf3aa15d7","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-08-16 00:58:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4e705b5022af7374d4cfe94bdd6368a64a12347a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"date":"2022-08-16 07:35:27.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"69ae0f425f7cbf94bb06727791cf94a2c4742c96","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 07:36:15.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"a9982a3e103965e9992ddf7ce4e9dfa76ffd6408","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 07:48:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/869ff533c27140f1878f1ac992047f5e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c79b0da4a3c94f00897ed1287caaeee9 : SUCCESS in 7m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/291fa6ed47be4e4a8dd3d9378895d447 : FAILURE in 5m 41s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c320f3fc789f4161b94e270e5c968e83 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":2},{"id":"e60f39d9dde86b0fe47827faa7c995d18c3e0f04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-08-16 08:02:21.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"4e95636452a51e96081f85d38dff60c616da48df","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-08-16 08:02:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c5213cd6c09d7efcc7cf3263ef4f28997e05d1fa","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-08-16 08:12:10.000000000","message":"Topic set to venus-deliverables-1st-release","accounts_in_message":[],"_revision_number":3},{"id":"3b6e02957fd9b1020bdcfe40b664ad3b39b5c153","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-08-16 08:12:43.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"496c222dc520d36f764c68dbaf3dd9f4f691c611","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 08:15:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3be10045b6ad45e49227f9cb6927ef58\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5386fb2f87c40cab8501d4a79c5a89d : SUCCESS in 7m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9c7a718b60da400c807087d118f595ed : SUCCESS in 5m 27s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/774bb4dfcabb4c56b9cc151cc1996bfc : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":3},{"id":"e561c1e731a4224b7fa1cb369b04b47138ea53a2","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-08-16 09:09:20.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3fc392a73964807545b107c4ab77eedd01971fdf","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-08-24 09:26:43.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"0348895161aca5be223c3aaeeabb65c2ea648e77","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-08-24 13:40:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3b71975595da4c7f1ce5b4cee3e3f3e312142bad","author":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"date":"2022-08-25 00:49:37.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"4e52c38e8ad804f4fc7be1348b60719c992cc55e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 00:50:27.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"0f93769bb223d28be932bb08e5561b8221ecf646","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-08-26 10:22:00.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e754a5bf52210ed9e06fcec8129520ebe40fc1e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 10:22:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ff5a3cd24d6cd94907ab023c44757ed874b8874b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 10:29:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a990fd31c2c540d7bbe1f05025bc0ffd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94dee2f1a1fc4bf3b19ff5155c6efc6a : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae81d828b97843818982e72e69ea93bd : SUCCESS in 6m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/451ee8673ac04c2b9c427f2d7ff89421 : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":3},{"id":"1321c063c3f0eb5814051134391e5effa2abb754","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 10:29:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"995baa3e24baa1170ec9a9d32bb6a21cf83662fc","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-08-26 10:51:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2a35d67921767ff478d27af5c197390bb52995ed","revisions":{"0dd17d226b55c8325bd206f79b7ce614a7b63f32":{"kind":"REWORK","_number":1,"created":"2022-08-15 10:17:00.000000000","uploader":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"ref":"refs/changes/58/853158/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/58/853158/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/853158/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/853158/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/853158/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/58/853158/1"}}},"commit":{"parents":[{"commit":"4502c0728987b7da8d0c1d2c4a538d8d78e59f6f","subject":"Merge \"cliff 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4502c0728987b7da8d0c1d2c4a538d8d78e59f6f"}]}],"author":{"name":"pangliye","email":"pangliye@inspur.com","date":"2022-08-15 10:15:28.000000000","tz":480},"committer":{"name":"pangliye","email":"pangliye@inspur.com","date":"2022-08-15 10:16:52.000000000","tz":480},"subject":"Release python-venusclient 1.0.0 for Zed","message":"Release python-venusclient 1.0.0 for Zed\n\nThis is the Zed milestone-3 release for python-venusclient,which is the first\nversion of python-venusclient.\n\nChange-Id: I62bb3606bed4aa3bf20f4cdb39a679bab89d7ce6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0dd17d226b55c8325bd206f79b7ce614a7b63f32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0dd17d226b55c8325bd206f79b7ce614a7b63f32"}]},"branch":"refs/heads/master"},"9167fd5770e29e4f056ec7e3618b60a240e71381":{"kind":"REWORK","_number":2,"created":"2022-08-16 07:35:27.000000000","uploader":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"ref":"refs/changes/58/853158/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/58/853158/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/853158/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/853158/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/853158/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/58/853158/2"}}},"commit":{"parents":[{"commit":"4502c0728987b7da8d0c1d2c4a538d8d78e59f6f","subject":"Merge \"cliff 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4502c0728987b7da8d0c1d2c4a538d8d78e59f6f"}]}],"author":{"name":"pangliye","email":"pangliye@inspur.com","date":"2022-08-15 10:15:28.000000000","tz":480},"committer":{"name":"pangliye","email":"pangliye@inspur.com","date":"2022-08-16 07:26:09.000000000","tz":0},"subject":"Release python-venusclient 1.0.0 for Zed","message":"Release python-venusclient 1.0.0 for Zed\n\nThis is the Zed milestone-3 release for python-venusclient,which is the first\nversion of python-venusclient.\n\nChange-Id: I62bb3606bed4aa3bf20f4cdb39a679bab89d7ce6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9167fd5770e29e4f056ec7e3618b60a240e71381"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9167fd5770e29e4f056ec7e3618b60a240e71381"}]},"branch":"refs/heads/master"},"2a35d67921767ff478d27af5c197390bb52995ed":{"kind":"REWORK","_number":3,"created":"2022-08-16 08:02:21.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/58/853158/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/58/853158/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/853158/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/853158/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/853158/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/58/853158/3"}}},"commit":{"parents":[{"commit":"4502c0728987b7da8d0c1d2c4a538d8d78e59f6f","subject":"Merge \"cliff 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4502c0728987b7da8d0c1d2c4a538d8d78e59f6f"}]}],"author":{"name":"pangliye","email":"pangliye@inspur.com","date":"2022-08-15 10:15:28.000000000","tz":480},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-08-16 08:02:05.000000000","tz":0},"subject":"Release python-venusclient 1.0.0 for Zed","message":"Release python-venusclient 1.0.0 for Zed\n\nThis is the Zed milestone-3 release for python-venusclient, which is\nthe first version of python-venusclient.\n\nChange-Id: I62bb3606bed4aa3bf20f4cdb39a679bab89d7ce6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2a35d67921767ff478d27af5c197390bb52995ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2a35d67921767ff478d27af5c197390bb52995ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[]}
