)]}'
{"id":"openstack%2Freleases~853475","triplet_id":"openstack%2Freleases~master~I8a2c389dbcb3e325f2a73ad86361481a406a0823","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2022-08-23 13:00:45.000000000","reason":"Change was marked work in progress"},"27427":{"account":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"last_update":"2022-08-22 12:53:14.000000000","reason":"\u003cGERRIT_ACCOUNT_27427\u003e replied on the change","reason_account":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"}},"16515":{"account":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"last_update":"2022-08-22 12:37:05.000000000","reason":"\u003cGERRIT_ACCOUNT_16515\u003e replied on the change","reason_account":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"}},"7144":{"account":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"last_update":"2022-08-23 13:00:45.000000000","reason":"Change was marked work in progress"},"32926":{"account":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"last_update":"2022-08-23 07:34:12.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_32926\u003e by clicking the attention icon","reason_account":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}}},"hashtags":[],"change_id":"I8a2c389dbcb3e325f2a73ad86361481a406a0823","subject":"Releasing validations-libs and validations-common 2.0.0","status":"ABANDONED","created":"2022-08-17 12:30:12.000000000","updated":"2023-02-02 10:42:15.000000000","total_comment_count":2,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"df23a85d1404bc51d4c75cef7cc2153e5fd71868","_number":853475,"virtual_id_number":853475,"owner":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-08-22 12:07:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2022-08-22 12:37:05.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}]},"reviewer_updates":[{"updated":"2022-08-17 12:31:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-17 12:54:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-17 13:20:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2022-08-17 13:20:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2022-08-18 14:11:39.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"state":"REVIEWER"},{"updated":"2022-08-18 14:11:39.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"state":"REVIEWER"},{"updated":"2022-09-22 15:27:02.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"}],"messages":[{"id":"b1e543e07bddadbb23ec69364d6de4c84883dfae","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-08-17 12:30:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35b61846b9009d8e21edbdaf13aaa0ca4e9b7d7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 12:54:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe742590d8af45f192dbfea6d697c7fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5d4050b3e1c4a87b1ad90006370b0e6 : SUCCESS in 7m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f52a6afa81be4516a2e6c62778e32458 : SUCCESS in 6m 38s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/beb548a0c5ad46a4a25a5641de810c93 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":1},{"id":"047277bc111c4cba6351cc2bc5b859a7499fb77f","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2022-08-19 14:05:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9090118b3c738d6070cb13f1d9b10ca99805f0f7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-08-22 11:51:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ec91464c4d133ec958fc2cb92a3870a859a6988d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 12:07:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/970750ee548045d89d3917ef0f40cae2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15d0fc7e363a47168bef8ff9eb442a61 : SUCCESS in 8m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/179d0f9289b0450ebac1ddd722479eac : SUCCESS in 8m 21s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e1579176061d48de9c9e326ce3c60110 : SUCCESS in 7m 21s","accounts_in_message":[],"_revision_number":2},{"id":"09e5aa0987f2a7f8dace7dcb0c015b77c73ec857","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-08-22 12:08:40.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"dd23899de8b8b35232b36f9fc74540913453f069","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2022-08-22 12:37:05.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a519953457f5afd2f483578849869dfdadaa6b3c","author":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"date":"2022-08-22 12:53:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ced360a40ef8956ab76fcacf2b8ad3eb1fe3c128","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-08-23 13:00:45.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"df23a85d1404bc51d4c75cef7cc2153e5fd71868","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-02 10:42:15.000000000","message":"Abandoned\n\nMeanwhile a similar patch was merged [1], that had newer hashes in it. So this one can be abandoned I think.\n\n[1] https://review.opendev.org/c/openstack/releases/+/863011","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f92aa281df3e27745d5c0b30b63f708ce59997ca","revisions":{"f57bc3e53e1653c31a1350451a2116fca958b21e":{"kind":"REWORK","_number":1,"created":"2022-08-17 12:30:12.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/75/853475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/75/853475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/853475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/853475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/853475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/75/853475/1"}}},"commit":{"parents":[{"commit":"4502c0728987b7da8d0c1d2c4a538d8d78e59f6f","subject":"Merge \"cliff 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4502c0728987b7da8d0c1d2c4a538d8d78e59f6f"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-08-17 12:26:20.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-08-17 12:26:20.000000000","tz":120},"subject":"Releasing validations-libs 2.0.0","message":"Releasing validations-libs 2.0.0\n\nMajor version incremented, due to removal of some previously supported\npython interpreter versions in 9047bbb317229a83b6855a22d4c943b6f9aaeada.\n\nvalidations-libs\u003e\u003d2.0.0 will only support python\u003e\u003d3.8\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I8a2c389dbcb3e325f2a73ad86361481a406a0823\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f57bc3e53e1653c31a1350451a2116fca958b21e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f57bc3e53e1653c31a1350451a2116fca958b21e"}]},"branch":"refs/heads/master"},"f92aa281df3e27745d5c0b30b63f708ce59997ca":{"kind":"REWORK","_number":2,"created":"2022-08-22 11:51:15.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/75/853475/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/75/853475/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/853475/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/853475/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/853475/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/75/853475/2"}}},"commit":{"parents":[{"commit":"4502c0728987b7da8d0c1d2c4a538d8d78e59f6f","subject":"Merge \"cliff 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4502c0728987b7da8d0c1d2c4a538d8d78e59f6f"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-08-17 12:26:20.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-08-22 11:47:28.000000000","tz":120},"subject":"Releasing validations-libs and validations-common 2.0.0","message":"Releasing validations-libs and validations-common 2.0.0\n\nMajor version incremented, due to removal of some previously supported\npython interpreter versions in 9047bbb317229a83b6855a22d4c943b6f9aaeada\nin validations-libs repo and analogous patch\n64f1e011e6e73bc524716d077d0c94015a3465e6 in validations-common.\n\nvalidations-libs\u003e\u003d2.0.0 will only support python\u003e\u003d3.8\nvalidations-commmon\u003e\u003d2.0.0 will only support python\u003e\u003d3.8\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I8a2c389dbcb3e325f2a73ad86361481a406a0823\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f92aa281df3e27745d5c0b30b63f708ce59997ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f92aa281df3e27745d5c0b30b63f708ce59997ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
