)]}'
{"id":"openstack%2Freleases~854956","triplet_id":"openstack%2Freleases~master~I3cfd845b8e82326df989eac90f291831e313a5c7","project":"openstack/releases","branch":"master","topic":"venus-deliverables-1st-release","attention_set":{},"removed_from_attention_set":{"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2022-09-01 08:17:03.000000000","reason":"\u003cGERRIT_ACCOUNT_308\u003e replied on the change","reason_account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-09-01 08:31:30.000000000","reason":"Change was submitted"},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2022-09-01 08:31:30.000000000","reason":"Change was submitted"},"24061":{"account":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"last_update":"2022-08-30 09:36:11.000000000","reason":"\u003cGERRIT_ACCOUNT_24061\u003e replied on the change","reason_account":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"}},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2022-09-01 08:31:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3cfd845b8e82326df989eac90f291831e313a5c7","subject":"Release 1.0.0.0b2 for venus","status":"MERGED","created":"2022-08-29 10:09:21.000000000","updated":"2022-09-01 09:29:57.000000000","submitted":"2022-09-01 08:31:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854956-venus-deliverables-1st-release","meta_rev_id":"d3900ec7e522d8eba0241cf3be930f11b7fcddac","_number":854956,"virtual_id_number":854956,"owner":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-01 08:31:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2022-09-01 08:17:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2022-08-30 09:41:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-30 09:36:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2022-09-01 08:17:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-08-30 09:37:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}],"CC":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-29 10:10:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-29 10:17:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-29 11:20:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"state":"REVIEWER"},{"updated":"2022-08-29 11:37:56.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2022-08-30 07:20:43.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2022-08-30 08:35:52.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"c944413bb1399af612932e5750f567a71ac5f08e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-08-29 10:09:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de55921b7d6ce5b75d61192b54ce294c51c2a5d7","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-08-29 10:10:56.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"82ae197045ebe5b70f85db50d68747e19bf8ff57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 10:17:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6dc35745dea547159da72a612e065184\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4920845189af40c2aac6205765781bfe : SUCCESS in 7m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5f6061dbc7544c5e9612f808df02343d : FAILURE in 6m 49s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a3d45a9cf96949bd8d441ae192422485 : FAILURE in 5m 26s","accounts_in_message":[],"_revision_number":1},{"id":"6d1fd002c5d174b0ed2c8908c4bed65b83d6cdc9","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-08-29 11:37:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2e510f9396c67034d6fd6b56f2eb1b94205ec85d","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-08-30 00:55:48.000000000","message":"Patch Set 1: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ebaee3c8dc9d9a302028d85c068e411e91b74665","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-08-30 00:59:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"895d075ccc23560d6b41b9b62b88db1b4f8b959b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 01:13:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e36ef1982b0644c89c4f75e4c5b22eb5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d540fbd550994179af6535d1d00f2ce7 : SUCCESS in 7m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7f267889a5b94c54b435a2433b7daf27 : FAILURE in 5m 18s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2c242957e4524b67b026b93d0005027e : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":1},{"id":"1b515a9f69652b1e697bb1cae79269e063a9c02f","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-08-30 02:33:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9a346932313ae19bd509f5a3740ea9d46d56b958","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 02:43:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c51cb9dd942435298a8529095060f79\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b599bd0c6cbf484ab95486f4d7f7bb0d : SUCCESS in 7m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5f8817fca6874017a7810b3a13022bbd : FAILURE in 6m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/bc362a7895244e4c99cf019734f57cbf : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":1},{"id":"032d5cf633482acc4cab70e410138ab56189e7d9","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-08-30 03:11:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"22f57a6ae338cd4aa3e995a5ff5afc10a4ccbd18","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-08-30 08:35:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9c75a00fb6f1a1472c85dbd1e1e985d547d97c15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-08-30 08:37:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a596b6691d62e5eb5d31f74ad35087fa179ba089","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 08:45:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba9f737e569e4d5a99ccad7132a3211e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3070bebd4b64a20815c487be5549b91 : SUCCESS in 6m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a458685008b2454d8dd19c43ceea7155 : FAILURE in 5m 43s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/67e132d0ab4646dcba074addeed8028b : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":2},{"id":"1e0959166faf2df6409d0435ab5ffd318e7644d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-08-30 09:17:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b434532c3c4fd709d06e9c39b8110663215761a2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-08-30 09:19:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"76a28cb9fae6bf7cc8517e0ca4731072909ca51f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 09:28:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc591a3937e0445ebd1560085aaeb7d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf13f3c1bfb7459a968c9e241fec3565 : SUCCESS in 7m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aad327a03c744150bf9a28e1a09d82d7 : SUCCESS in 6m 52s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/657929d73f0c4002855ea185a68db22d : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":3},{"id":"f86f22389ec318196cd6d243a20a49b57ec6ae58","author":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"date":"2022-08-30 09:36:11.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"be73870347af518cdf4ceecb7e8119730e5e86a2","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 09:37:11.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"dff999e232f7ffe12f9bc0ab1c98474faa8b3007","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-08-30 09:41:03.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"128fa1e7df1ff99d6f73bd9d1393636f5dc1cb8e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-09-01 08:17:03.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6a17f974154ceceb2152ee35bbf3a7e20ff5b597","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 08:17:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0e1d6bf36ec0cd1d8ab128f94d8cb8fa85e55aaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 08:31:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce908711214248239d88d416cb69bef6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/593e0d68c241488487b62b0985f83899 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e37d2aee8e84099a82b7c55fa365ef1 : SUCCESS in 7m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/51ecbd5f6c5f40eca9c54aa2f8b19ad8 : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":3},{"id":"498129006ec53d29571f476360190afe0def0c27","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 08:31:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d3900ec7e522d8eba0241cf3be930f11b7fcddac","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-01 09:29:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"adfd225aa09341f0b1a23b074d6ad2ba2b273800","revisions":{"4e44f674d6c41e12792855970eb6a2070cffa17a":{"kind":"REWORK","_number":1,"created":"2022-08-29 10:09:21.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/56/854956/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/56/854956/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/854956/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/854956/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/854956/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/56/854956/1"}}},"commit":{"parents":[{"commit":"56dafbc2e296f6f6581232021b653a627d1f2689","subject":"Merge \"Release final mistral-extra for Zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/56dafbc2e296f6f6581232021b653a627d1f2689"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2022-08-29 10:06:04.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2022-08-29 10:06:13.000000000","tz":480},"subject":"Release 1.0.0.0b2 for venus","message":"Release 1.0.0.0b2 for venus\n\nDue to the non-standard naming of the venus setup.cf directory,\npypi cannot generate \"openstack-venus\", fix the related problems\nand release the 1.0.0.0b2 version.\n\nChange-Id: I3cfd845b8e82326df989eac90f291831e313a5c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4e44f674d6c41e12792855970eb6a2070cffa17a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4e44f674d6c41e12792855970eb6a2070cffa17a"}]},"branch":"refs/heads/master"},"1791aa2ca34109955fb0e8a7270c2bc0b9f32d31":{"kind":"REWORK","_number":2,"created":"2022-08-30 08:37:11.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/56/854956/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/56/854956/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/854956/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/854956/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/854956/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/56/854956/2"}}},"commit":{"parents":[{"commit":"56dafbc2e296f6f6581232021b653a627d1f2689","subject":"Merge \"Release final mistral-extra for Zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/56dafbc2e296f6f6581232021b653a627d1f2689"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2022-08-29 10:06:04.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2022-08-30 08:36:27.000000000","tz":480},"subject":"Release 1.0.0.0b2 for venus","message":"Release 1.0.0.0b2 for venus\n\nDue to the non-standard naming of the venus setup.cf directory,\npypi cannot generate \"openstack-venus\", fix the related problems\nand release the 1.0.0.0b2 version.\n\nChange-Id: I3cfd845b8e82326df989eac90f291831e313a5c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1791aa2ca34109955fb0e8a7270c2bc0b9f32d31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1791aa2ca34109955fb0e8a7270c2bc0b9f32d31"}]},"branch":"refs/heads/master"},"adfd225aa09341f0b1a23b074d6ad2ba2b273800":{"kind":"REWORK","_number":3,"created":"2022-08-30 09:17:18.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/56/854956/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/56/854956/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/854956/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/854956/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/56/854956/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/56/854956/3"}}},"commit":{"parents":[{"commit":"56dafbc2e296f6f6581232021b653a627d1f2689","subject":"Merge \"Release final mistral-extra for Zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/56dafbc2e296f6f6581232021b653a627d1f2689"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2022-08-29 10:06:04.000000000","tz":480},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-08-30 09:15:50.000000000","tz":120},"subject":"Release 1.0.0.0b2 for venus","message":"Release 1.0.0.0b2 for venus\n\nDue to the non-standard naming of the venus setup.cf directory,\npypi cannot generate \"openstack-venus\", fix the related problems\nand release the 1.0.0.0b2 version.\n\nChange-Id: I3cfd845b8e82326df989eac90f291831e313a5c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/adfd225aa09341f0b1a23b074d6ad2ba2b273800"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/adfd225aa09341f0b1a23b074d6ad2ba2b273800"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[]}
