)]}'
{"id":"openstack%2Freleases~857178","triplet_id":"openstack%2Freleases~master~Ifeea33a6bf299fe7acce4f5c90db98ba2d301ade","project":"openstack/releases","branch":"master","topic":"zed-rc1-deadline","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-09-16 15:01:26.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"6732":{"account":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"last_update":"2022-09-16 15:14:44.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2022-09-16 15:14:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifeea33a6bf299fe7acce4f5c90db98ba2d301ade","subject":"Release trove RC1 for Zed","status":"MERGED","created":"2022-09-12 13:30:27.000000000","updated":"2022-09-22 14:50:53.000000000","submitted":"2022-09-16 15:14:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"857178-zed-rc1-deadline","meta_rev_id":"405a58e627d6c1fdb0d80d1158617ff21e7f4833","_number":857178,"virtual_id_number":857178,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-16 15:14:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2022-09-16 09:44:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2022-09-16 10:53:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2022-09-16 15:01:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-12 13:31:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-12 13:44:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-12 14:20:31.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2022-09-12 14:20:31.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2022-09-16 10:53:14.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"da01d26d282b8b8eb4d676b05288a90c603cfebc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-12 13:30:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da04a533c728b26621a2d79e16adbdf26cd58dec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 13:44:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a187ab2308d845caac1692d277e0cfde\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91f18558e9464c7facc82bf8966dc541 : SUCCESS in 7m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/91bf005575e64910bcf42d3b7bf09b5d : SUCCESS in 5m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/117955ac567341a982ece9b566b8c7f8 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"ce7d2bea15e15fafec56f38b0b409edbc7a903bb","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-09-13 05:51:14.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"38c4e5a1ad5ee4f425c9482f1298eabd6ca7a838","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-09-15 10:12:14.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e358b291436ad843bbacf13174b7f5e37abd092c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-16 09:38:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0ac67933591f2f3e34f26ff1311fcd55144bc673","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-16 09:44:16.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"df90d54424bd45fa1e7bea1b6eb723da70e9f4ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 09:52:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b92fc48ce3f4dceb94dec37bd0f6947\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36c3149a50a947f484629692aef6984f : SUCCESS in 7m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/27c06f14f063441c9b63f3b3650b4647 : SUCCESS in 5m 14s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d3a7e97fb38e4faeaac26d4b705b431f : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":2},{"id":"47a9dd4080b298606e7c7ef73af3c3e27461cec3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-09-16 10:53:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0eb11a37e3005e70c5d14859a06d98904f9a0eef","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-16 15:01:26.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a6404733078fbbaa18ce03188ef8ed58e86a5f53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 15:02:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"85e32d72ac020a3c35f811b95f8767d22c3222b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 15:14:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b1cf8dfe77a47068e97252ec93920a1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5661e03b4b964a419557c8e9541957be : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b09a59e3f614438b92c2044142766dc : SUCCESS in 10m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a9d8931f054f4c7ab2ae0184ab973f18 : SUCCESS in 8m 24s","accounts_in_message":[],"_revision_number":2},{"id":"e41d3aa40934fc1ca1d238d6b56f6aa229596ca6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 15:14:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ea959b34d250169918125df7cc0546b6510596de","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-09-20 03:38:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"405a58e627d6c1fdb0d80d1158617ff21e7f4833","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-22 14:50:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c75828361712202cac5556b7607813243a345e83","revisions":{"87a257d6f86efc4762ee5e2c6debf2506b4b54d3":{"kind":"REWORK","_number":1,"created":"2022-09-12 13:30:27.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/78/857178/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/78/857178/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/857178/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/857178/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/857178/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/78/857178/1"}}},"commit":{"parents":[{"commit":"6801b60d1e622d19ff14d006cbd49c829c2d8af0","subject":"Merge \"[trove] Create zed branch for client and non-client libs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6801b60d1e622d19ff14d006cbd49c829c2d8af0"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-12 13:30:23.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-12 13:30:23.000000000","tz":120},"subject":"Release trove RC1 for Zed","message":"Release trove RC1 for Zed\n\nThis proposed a RC1 release for trove to get any merged changes\nout prior to the RC1 deadline for Zed going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, September 16.\n\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ifeea33a6bf299fe7acce4f5c90db98ba2d301ade\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/87a257d6f86efc4762ee5e2c6debf2506b4b54d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/87a257d6f86efc4762ee5e2c6debf2506b4b54d3"}]},"branch":"refs/heads/master"},"c75828361712202cac5556b7607813243a345e83":{"kind":"REWORK","_number":2,"created":"2022-09-16 09:38:49.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/78/857178/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/78/857178/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/857178/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/857178/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/78/857178/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/78/857178/2"}}},"commit":{"parents":[{"commit":"6801b60d1e622d19ff14d006cbd49c829c2d8af0","subject":"Merge \"[trove] Create zed branch for client and non-client libs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6801b60d1e622d19ff14d006cbd49c829c2d8af0"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-12 13:30:23.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-09-16 09:38:21.000000000","tz":120},"subject":"Release trove RC1 for Zed","message":"Release trove RC1 for Zed\n\nThis proposed a RC1 release for trove to get any merged changes\nout prior to the RC1 deadline for Zed going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, September 16.\n\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ifeea33a6bf299fe7acce4f5c90db98ba2d301ade\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c75828361712202cac5556b7607813243a345e83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c75828361712202cac5556b7607813243a345e83"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[]}
