)]}'
{"id":"openstack%2Freleases~857180","triplet_id":"openstack%2Freleases~master~I71dfc3c331a3e4e385c4190d1ff05a1281dc5acb","project":"openstack/releases","branch":"master","topic":"zed-rc1-deadline","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-09-16 12:30:21.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2022-09-16 12:37:32.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-09-16 12:20:02.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},"24061":{"account":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"last_update":"2022-09-16 12:37:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I71dfc3c331a3e4e385c4190d1ff05a1281dc5acb","subject":"Release venus RC1 for Zed","status":"MERGED","created":"2022-09-12 13:31:23.000000000","updated":"2022-09-16 12:37:32.000000000","submitted":"2022-09-16 12:37:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"857180-zed-rc1-deadline","meta_rev_id":"364b2f6242bbac51b1af6b28feeb9917fda613b6","_number":857180,"virtual_id_number":857180,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-16 12:37:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2022-09-16 12:30:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":2,"date":"2022-09-16 12:20:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2022-09-16 12:30:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"}],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-12 13:32:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-12 13:39:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-12 14:21:00.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"state":"REVIEWER"},{"updated":"2022-09-12 14:21:01.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2022-09-15 08:29:05.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"00a0526fa8c912e35773b5d68ba58bd31c8b6964","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-12 13:31:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0725945b9bb2d1d8e27c8cd8aab65a1d3cb638e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 13:39:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a980e6eb18c54a45bb9a9d1eebb518b2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cde2d6ab3854db8b978a7af56c249e2 : SUCCESS in 6m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/60ddb4344d4843929b08b6c65a58b4cd : SUCCESS in 5m 53s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/64a12f31c8a04a219fa13eb4e9341b93 : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":1},{"id":"ce4990cc8f008d4942cdc8a2da7343b639ddf14c","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-09-14 05:38:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cb1e51904e722e7555f2ec8db78fed1218617130","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-09-15 00:50:47.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"cc0096c08525e2528a819c9931f6776b0bd7bc34","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-09-15 00:50:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f435f3e75df88584865c368d03ed3795e021aca0","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 00:51:36.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"b97d8300c168aed4112acc1caf5047c545fceafe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 01:02:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d81be0aada4e4fa18ec3ee41b581ee28\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dde23924922450d9b81747f9ce71da2 : SUCCESS in 6m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fbf069837c2d483989ffd45463da959c : SUCCESS in 5m 06s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3607b5a6c2914a0982c4c352b552bf6c : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":2},{"id":"a84b1623a4be5882fddb01ee8053f9272bcbab84","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-09-15 08:29:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"48325238cad0c39764988f24fcd5b828ce496037","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-15 08:37:29.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3bd92a9c8e5cd88acc584f6e78f4e482f60087a6","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-09-16 00:41:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ea1d34c556a941998e9d56b242be705e5460f6ad","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-09-16 07:45:59.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"b657a46bfe64c622880311fdf79ba78573bbcdec","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-09-16 07:48:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"07c7501c9b77e5ab1db9d8a1c02c236de2eb272f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-16 11:36:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d0254cfd40e6fe707b326859a2d832d1281d4251","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-16 11:47:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d67858f96276b98a6576dff3f61552fa65ed6c99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 11:49:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d5ed70ad129459e834146d5d7b6e5fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29627898bb4d421f89692fad5794bd9f : SUCCESS in 6m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2a1c9fe390c1487bbd072ffcb4cfc0ca : SUCCESS in 5m 08s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b67a3180b0734b8bb883040b12ec171b : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":3},{"id":"e6b3f54b538426fc889211d577d0781806437591","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-09-16 12:20:02.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"af7fa30241cabcd5c0cb5cf35ebc22771fdfc1c7","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-16 12:30:21.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b316fd64cfb08e3602f174ef7fbfdb63101461ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 12:30:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"77b028c9c6fb57977a71118ff8779cccdf29ccf7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 12:37:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db0289f4fcf94c279599d54414094b5f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fa43bced64d4502a2d175dfe4a930d8 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba3234af35984a6688709903f672fbaf : SUCCESS in 6m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8e7634f28256417f898f3e48a3a12354 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":3},{"id":"364b2f6242bbac51b1af6b28feeb9917fda613b6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 12:37:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e1606bd60a8a921294c7f1b569d389a8cd05e8ea","revisions":{"e320fbfb61afde443ac52fdeaa6c86c4b39746cf":{"kind":"REWORK","_number":1,"created":"2022-09-12 13:31:23.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/80/857180/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/80/857180/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/857180/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/857180/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/857180/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/80/857180/1"}}},"commit":{"parents":[{"commit":"6801b60d1e622d19ff14d006cbd49c829c2d8af0","subject":"Merge \"[trove] Create zed branch for client and non-client libs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6801b60d1e622d19ff14d006cbd49c829c2d8af0"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-12 13:31:19.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-12 13:31:19.000000000","tz":120},"subject":"Release venus RC1 for Zed","message":"Release venus RC1 for Zed\n\nThis proposed a RC1 release for venus to get any merged changes\nout prior to the RC1 deadline for Zed going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, September 16.\n\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I71dfc3c331a3e4e385c4190d1ff05a1281dc5acb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e320fbfb61afde443ac52fdeaa6c86c4b39746cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e320fbfb61afde443ac52fdeaa6c86c4b39746cf"}]},"branch":"refs/heads/master"},"8fb528d04d01af7b04bc29691062a6e963b6f1cf":{"kind":"REWORK","_number":2,"created":"2022-09-15 00:50:47.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/80/857180/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/80/857180/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/857180/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/857180/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/857180/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/80/857180/2"}}},"commit":{"parents":[{"commit":"6801b60d1e622d19ff14d006cbd49c829c2d8af0","subject":"Merge \"[trove] Create zed branch for client and non-client libs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6801b60d1e622d19ff14d006cbd49c829c2d8af0"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-12 13:31:19.000000000","tz":120},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2022-09-15 00:50:37.000000000","tz":0},"subject":"Release venus RC1 for Zed","message":"Release venus RC1 for Zed\n\nThis proposed a RC1 release for venus to get any merged changes\nout prior to the RC1 deadline for Zed going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, September 16.\n\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I71dfc3c331a3e4e385c4190d1ff05a1281dc5acb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8fb528d04d01af7b04bc29691062a6e963b6f1cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8fb528d04d01af7b04bc29691062a6e963b6f1cf"}]},"branch":"refs/heads/master"},"e1606bd60a8a921294c7f1b569d389a8cd05e8ea":{"kind":"REWORK","_number":3,"created":"2022-09-16 11:36:34.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/80/857180/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/80/857180/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/857180/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/857180/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/857180/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/80/857180/3"}}},"commit":{"parents":[{"commit":"6801b60d1e622d19ff14d006cbd49c829c2d8af0","subject":"Merge \"[trove] Create zed branch for client and non-client libs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6801b60d1e622d19ff14d006cbd49c829c2d8af0"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-12 13:31:19.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-09-16 11:36:02.000000000","tz":120},"subject":"Release venus RC1 for Zed","message":"Release venus RC1 for Zed\n\nThis proposed a RC1 release for venus to get any merged changes\nout prior to the RC1 deadline for Zed going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, September 16.\n\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I71dfc3c331a3e4e385c4190d1ff05a1281dc5acb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e1606bd60a8a921294c7f1b569d389a8cd05e8ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e1606bd60a8a921294c7f1b569d389a8cd05e8ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[]}
