)]}'
{"id":"openstack%2Freleases~858415","triplet_id":"openstack%2Freleases~master~I21d72e673e9edd83d98a0b3010fb2828d79ddd94","project":"openstack/releases","branch":"master","topic":"zed-tp-latest","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-09-23 16:19:45.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-09-23 16:19:45.000000000","reason":"Change was submitted"},"9003":{"account":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"last_update":"2022-09-23 16:19:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I21d72e673e9edd83d98a0b3010fb2828d79ddd94","subject":"Release manila-tempest-plugin for Zed","status":"MERGED","created":"2022-09-19 16:10:35.000000000","updated":"2022-09-23 16:19:45.000000000","submitted":"2022-09-23 16:19:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858415-zed-tp-latest","meta_rev_id":"5c2683c164fd17b29cec94f93dd3d5a53f03da74","_number":858415,"virtual_id_number":858415,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-23 16:19:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2022-09-23 16:00:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2022-09-23 16:00:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"}],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-19 16:11:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-19 16:20:30.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2022-09-19 16:20:31.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-09-19 16:20:32.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2022-09-19 16:34:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bace710a7432cdeca300b8c5d8621f7ba18ba45c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-19 16:10:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ebe86c6a02012ea186d0348325759574341ad3b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 16:34:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4beee8dabd034ee1b2edf69cbebecb00\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6ee4faffa47404f935224302fd16d8b : SUCCESS in 7m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5d035691793044b79f3da287eed5bcc9 : SUCCESS in 6m 27s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/317fde07bb5948d88a07527e28e09e89 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":1},{"id":"17ee3b38890db933ededb56f3e749e9112cb7d39","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-23 16:00:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d3a9c011e66e77394daa2414bab76a3ef98af531","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 16:01:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1254efafcfe51f4531ed89191a33639778531a3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 16:19:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af5ca59f53ea4a7c8891b21902290d42\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66687a7115ac44f9a98aca3c252134a8 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65249cf58f8d4bdc9f1288017a36dc13 : SUCCESS in 7m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/063ed9fdb95b4b928f5fbd8016e0de07 : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":1},{"id":"5c2683c164fd17b29cec94f93dd3d5a53f03da74","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 16:19:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e439e79bde89599b79cd2d4578fbc058b8f87080","revisions":{"e439e79bde89599b79cd2d4578fbc058b8f87080":{"kind":"REWORK","_number":1,"created":"2022-09-19 16:10:35.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/15/858415/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/15/858415/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/858415/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/858415/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/858415/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/15/858415/1"}}},"commit":{"parents":[{"commit":"faf12d5b0a7a80b40424b7d5fe4d90d89a4e0825","subject":"Merge \"Creating stable zed for vitrage-dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/faf12d5b0a7a80b40424b7d5fe4d90d89a4e0825"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-19 16:10:31.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-19 16:10:31.000000000","tz":120},"subject":"Release manila-tempest-plugin for Zed","message":"Release manila-tempest-plugin for Zed\n\nThis patch proposes a release for manila-tempest-plugin\nto get the latest merged changes out prior to the final Zed going\ninto effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, September 23.\n\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I21d72e673e9edd83d98a0b3010fb2828d79ddd94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e439e79bde89599b79cd2d4578fbc058b8f87080"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e439e79bde89599b79cd2d4578fbc058b8f87080"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[]}
