)]}'
{"id":"openstack%2Freleases~858419","triplet_id":"openstack%2Freleases~master~I6686756889dc71ae07683c169aa3094c66dc5e6b","project":"openstack/releases","branch":"master","topic":"zed-tp-latest","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-09-26 12:02:33.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-09-26 14:37:45.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-09-26 14:37:45.000000000","reason":"Change was submitted"},"841":{"account":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"last_update":"2022-09-26 14:37:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6686756889dc71ae07683c169aa3094c66dc5e6b","subject":"Release neutron-tempest-plugin for Zed","status":"MERGED","created":"2022-09-19 16:19:47.000000000","updated":"2022-09-26 14:37:45.000000000","submitted":"2022-09-26 14:37:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"858419-zed-tp-latest","meta_rev_id":"c7f24ee1d0dd709bff83e83715d8f9202a0d698e","_number":858419,"virtual_id_number":858419,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-26 14:37:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-09-26 12:02:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-09-26 10:34:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2022-09-26 14:27:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2022-09-26 14:27:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-09-26 12:03:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-19 16:20:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-09-19 16:20:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2022-09-19 16:21:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-19 16:48:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-22 16:25:27.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-09-26 14:27:44.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"41dc5435067637de9b47b1fc8b39db7c24403e6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-19 16:19:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d3ba25af7b607ad6b80501f3f3345bf280c0b14","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-09-19 16:27:41.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"36ad209a2d0ee7e3892c6c02f24879df9cd893a0","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-19 16:36:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5445fa62bb6b509c0071da91c712a06407fca1a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 16:48:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8df1b46d23dd470cbf4dece6f3ccfadf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ccaf8b01c4b47a2be01423c6d1124b5 : SUCCESS in 7m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1c87cb83a3834b89a82e3b08d52c9c0b : SUCCESS in 6m 33s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/78ebeb63466443a989a8ae0cbce131eb : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":1},{"id":"2c26fb1845de6b569ef9ce5bfe836923a22ae0bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-09-25 07:55:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a4aab1a4ce39788abd4eac7e01ad912ded768fb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-25 08:04:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4c8e69d61704620b441b3e47dbf187d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41a5dada80c44e81a0e9b0d6bd2ff501 : SUCCESS in 8m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e1864234d76d4eeebb6b1dd694ff41aa : FAILURE in 6m 57s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0555e370b2da4feb901c1100f98de330 : FAILURE in 5m 47s","accounts_in_message":[],"_revision_number":2},{"id":"af070d146267c1a40fa59622e453a417eee61c33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-09-26 10:03:46.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"051d516764f6b4c20572e50eda1462fddfa09509","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 10:17:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a008fbf69c764bfa82837ba72ae28f3f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d2122ec53a043369041c65141f594e3 : SUCCESS in 7m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/45329fccf9644cbeb5c0f6e48b6dd9e9 : SUCCESS in 6m 04s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/73fff0d2dc104cef816b3c9576475637 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":3},{"id":"0d0ff2c1471204d597f0f7096897ef638b31f365","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-26 10:34:31.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0f19315a627b7ec9f51a92c33d4a654ebcf7affa","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-09-26 12:02:33.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"cc56112fc30efbc0f0b3e64c25e7d02752a1421e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 12:03:31.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"f67748ef4e83769e436c1ffb5c04d8688d74c94a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-09-26 14:27:44.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9e3d91adaa44702595af8c07e0201d6a70ae767e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 14:28:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"748f0f54d4c3c57d6c544c030ed25c85572a2ed6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 14:37:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/160412c9d94a4bd8b12d065934aa1837\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfed23cfffeb43889ff18a003c41de37 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43d13092f7654ef5babb24dbeedbd358 : SUCCESS in 7m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f245669d25e84c73b2d36e3adb7d4f6a : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":3},{"id":"c7f24ee1d0dd709bff83e83715d8f9202a0d698e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 14:37:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"516e803795303dff9548537b96311144e9fa3cfe","revisions":{"0434dc61f42074bff97af9c4bc861406f7b6a26b":{"kind":"REWORK","_number":1,"created":"2022-09-19 16:19:47.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/19/858419/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/19/858419/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/858419/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/858419/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/858419/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/19/858419/1"}}},"commit":{"parents":[{"commit":"faf12d5b0a7a80b40424b7d5fe4d90d89a4e0825","subject":"Merge \"Creating stable zed for vitrage-dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/faf12d5b0a7a80b40424b7d5fe4d90d89a4e0825"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-19 16:19:43.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-19 16:19:43.000000000","tz":120},"subject":"Release neutron-tempest-plugin for Zed","message":"Release neutron-tempest-plugin for Zed\n\nThis patch proposes a release for neutron-tempest-plugin\nto get the latest merged changes out prior to the final Zed going\ninto effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, September 23.\n\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I6686756889dc71ae07683c169aa3094c66dc5e6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0434dc61f42074bff97af9c4bc861406f7b6a26b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0434dc61f42074bff97af9c4bc861406f7b6a26b"}]},"branch":"refs/heads/master"},"2b2f4230ef48f5a87c16744acba18582d0a3d880":{"kind":"REWORK","_number":2,"created":"2022-09-25 07:55:33.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/19/858419/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/19/858419/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/858419/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/858419/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/858419/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/19/858419/2"}}},"commit":{"parents":[{"commit":"faf12d5b0a7a80b40424b7d5fe4d90d89a4e0825","subject":"Merge \"Creating stable zed for vitrage-dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/faf12d5b0a7a80b40424b7d5fe4d90d89a4e0825"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-19 16:19:43.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-09-25 07:55:16.000000000","tz":120},"subject":"Release neutron-tempest-plugin for Zed","message":"Release neutron-tempest-plugin for Zed\n\nThis patch proposes a release for neutron-tempest-plugin\nto get the latest merged changes out prior to the final Zed going\ninto effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, September 23.\n\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I6686756889dc71ae07683c169aa3094c66dc5e6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2b2f4230ef48f5a87c16744acba18582d0a3d880"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2b2f4230ef48f5a87c16744acba18582d0a3d880"}]},"branch":"refs/heads/master"},"516e803795303dff9548537b96311144e9fa3cfe":{"kind":"REWORK","_number":3,"created":"2022-09-26 10:03:46.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/19/858419/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/19/858419/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/858419/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/858419/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/19/858419/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/19/858419/3"}}},"commit":{"parents":[{"commit":"faf12d5b0a7a80b40424b7d5fe4d90d89a4e0825","subject":"Merge \"Creating stable zed for vitrage-dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/faf12d5b0a7a80b40424b7d5fe4d90d89a4e0825"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-19 16:19:43.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2022-09-26 10:03:46.000000000","tz":0},"subject":"Release neutron-tempest-plugin for Zed","message":"Release neutron-tempest-plugin for Zed\n\nThis patch proposes a release for neutron-tempest-plugin\nto get the latest merged changes out prior to the final Zed going\ninto effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by early Friday, September 23.\n\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I6686756889dc71ae07683c169aa3094c66dc5e6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/516e803795303dff9548537b96311144e9fa3cfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/516e803795303dff9548537b96311144e9fa3cfe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[]}
