)]}'
{"id":"openstack%2Freleases~858488","triplet_id":"openstack%2Freleases~master~If370a7968e94f4487812e19c64f8cb8a9156dcb3","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2022-09-20 09:47:36.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-09-20 09:31:56.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-09-20 08:45:39.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}},"hashtags":[],"change_id":"If370a7968e94f4487812e19c64f8cb8a9156dcb3","subject":"Release horizon 23.0.0(Zed)","status":"MERGED","created":"2022-09-20 06:53:26.000000000","updated":"2022-09-20 09:47:36.000000000","submitted":"2022-09-20 09:47:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858488","meta_rev_id":"747749df0afed3a954d513e6fa7d6a32829c1f83","_number":858488,"virtual_id_number":858488,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-20 09:47:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2022-09-20 09:31:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2022-09-20 09:34:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2022-09-20 09:34:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-09-20 08:51:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-20 06:53:45.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2022-09-20 06:54:22.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-09-20 06:54:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a86552416522802ebe37312606321ecfb0fc44c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-09-20 06:53:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"63d2ddd8662d9de23ecaad1cf4034674c1a61561","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 06:54:27.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"b11cca5fc894c8809a9b332098cd72044be1a139","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 07:02:12.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/948d4930bf0b4bf3b6fe00b13ae74222\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d126ca66fee24d359cc61026cb9561be : SUCCESS in 7m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8e4f362529b3490bbc6eece1bb328c70 : FAILURE in 4m 13s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f76df4f5d4024e45868c97b5d931e936 : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":1},{"id":"ed55f4c816de0ff7ec0cb1339a36c34e6b12b58d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-09-20 07:06:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bb45c45c2b1c7b18a414f4003620df20b5808f8f","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 07:06:57.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"f84a282ef9824970e0cc0ec760f16b36e1291c3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 07:14:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fb5a9ff0cfa4e9a9b3b33279223c3a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b83cb018e4e4a64b29ac43818342aca : SUCCESS in 6m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/54e0de3b9c7b4737b8ce26b511702ab4 : SUCCESS in 6m 00s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/75a0ffb8e8a4479c8218023c421064cf : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":2},{"id":"2b28fca22b68d3c92aab792ff46478577bdce484","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-20 07:20:22.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d0010df0182beb324c56347a3d3430edfd0378c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-09-20 08:39:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"07f50b58fc68f7bd2b06c7958a711da851c2653d","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 08:40:47.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"5a5579d6684dbb12ed9ca9503b4e9a1d2ab00d71","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-09-20 08:43:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d5dfaa518e6da2a9de31e2a519f33a660a090f63","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-09-20 08:45:39.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"735ff3d259406891703d8321e744b380ab270a6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-09-20 08:50:23.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"914ccc8661fdafede969a62ae0b660c878dd0904","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-09-20 08:50:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ac8de233943e4a5c2e69f4d6d3bf48c54d16fc27","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 08:51:13.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"88d81ea1e5b82c0f02f6b09d11045c911b445216","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 09:05:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80d937221e4749089f6135e6a82e6e72\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bac263c5aa7441abeb8af5e6123b6ff : SUCCESS in 6m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4e7f588b3a68469fb1532ad13a82c064 : SUCCESS in 6m 07s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d80c180b0ded4951bd696a24cce4fbd3 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":4},{"id":"3c62582caf44ef9a7caea5fbd3c8ea3ceefd05de","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-20 09:31:56.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7fe136df92fda957d8cf5af75a3d976db572bae8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-09-20 09:34:18.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f267e5faaf68eda7945d0fee1bba2aa90eff257b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 09:35:08.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1f7afafbb1553e733ec18a44eb90aa374e6f7bfb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 09:47:36.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ab3db8ea2094560a583a04695b6a8d4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48e7610d376d44c684fa9d0da41b26cc : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2851fd3d796247298dc69ebd0bb34529 : SUCCESS in 6m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2c08d538306d4f86b0511485c08bff9c : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":4},{"id":"747749df0afed3a954d513e6fa7d6a32829c1f83","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 09:47:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ab2b58e16130b573962e1c563d292e289b8cb0bf","revisions":{"6c13f20e22733d5dae8b6b9b5f52a7c3711983df":{"kind":"REWORK","_number":1,"created":"2022-09-20 06:53:26.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/88/858488/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/88/858488/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/858488/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/858488/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/858488/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/88/858488/1"}}},"commit":{"parents":[{"commit":"faf12d5b0a7a80b40424b7d5fe4d90d89a4e0825","subject":"Merge \"Creating stable zed for vitrage-dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/faf12d5b0a7a80b40424b7d5fe4d90d89a4e0825"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-09-20 06:00:55.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-09-20 06:07:53.000000000","tz":330},"subject":"Release horizon 22.3.0(Zed)","message":"Release horizon 22.3.0(Zed)\n\nThis patch cut a new release of the horizon which can\nbe used to create stable/zed branch as there are many\nimportant changes feature/bug fixes are merged in horizon\nsince the last release.\n\nChange-Id: If370a7968e94f4487812e19c64f8cb8a9156dcb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6c13f20e22733d5dae8b6b9b5f52a7c3711983df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6c13f20e22733d5dae8b6b9b5f52a7c3711983df"}]},"branch":"refs/heads/master"},"95885a18fce63ee66d0cd7b280dd47a8f623430c":{"kind":"REWORK","_number":2,"created":"2022-09-20 07:06:12.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/88/858488/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/88/858488/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/858488/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/858488/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/858488/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/88/858488/2"}}},"commit":{"parents":[{"commit":"faf12d5b0a7a80b40424b7d5fe4d90d89a4e0825","subject":"Merge \"Creating stable zed for vitrage-dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/faf12d5b0a7a80b40424b7d5fe4d90d89a4e0825"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-09-20 06:00:55.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-09-20 06:18:00.000000000","tz":330},"subject":"Release horizon 22.3.0(Zed)","message":"Release horizon 22.3.0(Zed)\n\nThis patch cut a new release of the horizon which can\nbe used to create stable/zed branch as there are many\nimportant changes feature/bug fixes are merged in horizon\nsince the last release.\n\nChange-Id: If370a7968e94f4487812e19c64f8cb8a9156dcb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/95885a18fce63ee66d0cd7b280dd47a8f623430c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/95885a18fce63ee66d0cd7b280dd47a8f623430c"}]},"branch":"refs/heads/master"},"1c9669277c7aa31148701e76e36862a773e1936b":{"kind":"REWORK","_number":3,"created":"2022-09-20 08:39:46.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/88/858488/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/88/858488/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/858488/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/858488/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/858488/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/88/858488/3"}}},"commit":{"parents":[{"commit":"faf12d5b0a7a80b40424b7d5fe4d90d89a4e0825","subject":"Merge \"Creating stable zed for vitrage-dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/faf12d5b0a7a80b40424b7d5fe4d90d89a4e0825"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-09-20 06:00:55.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-09-20 08:38:56.000000000","tz":330},"subject":"Release horizon 22.3.0(Zed)","message":"Release horizon 22.3.0(Zed)\n\nThis patch cut a new release of the horizon which can\nbe used to create stable/zed branch as there are many\nimportant changes feature/bug fixes are merged in horizon\nsince the last release.\n\nChange-Id: If370a7968e94f4487812e19c64f8cb8a9156dcb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1c9669277c7aa31148701e76e36862a773e1936b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1c9669277c7aa31148701e76e36862a773e1936b"}]},"branch":"refs/heads/master"},"ab2b58e16130b573962e1c563d292e289b8cb0bf":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-09-20 08:50:23.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/88/858488/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/88/858488/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/858488/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/858488/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/858488/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/88/858488/4"}}},"commit":{"parents":[{"commit":"faf12d5b0a7a80b40424b7d5fe4d90d89a4e0825","subject":"Merge \"Creating stable zed for vitrage-dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/faf12d5b0a7a80b40424b7d5fe4d90d89a4e0825"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-09-20 06:00:55.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-09-20 08:48:33.000000000","tz":330},"subject":"Release horizon 23.0.0(Zed)","message":"Release horizon 23.0.0(Zed)\n\nThis patch cut a new release of the horizon which can\nbe used to create stable/zed branch as there are many\nimportant changes feature/bug fixes are merged in horizon\nsince the last release.\n\nChange-Id: If370a7968e94f4487812e19c64f8cb8a9156dcb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ab2b58e16130b573962e1c563d292e289b8cb0bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ab2b58e16130b573962e1c563d292e289b8cb0bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[]}
