)]}'
{"id":"openstack%2Freleases~860029","triplet_id":"openstack%2Freleases~master~I65d51978ab978b66d6d67c5c323b29dc59c62de8","project":"openstack/releases","branch":"master","topic":"zed-final","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-10-05 14:28:34.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}},"hashtags":[],"change_id":"I65d51978ab978b66d6d67c5c323b29dc59c62de8","subject":"Mark Zed as released","status":"MERGED","created":"2022-09-30 19:25:44.000000000","updated":"2023-03-22 14:11:56.000000000","submitted":"2022-10-05 14:40:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860029-zed-final","meta_rev_id":"f86a3f2ce55288c04ee762d7e4f21872cf59779f","_number":860029,"virtual_id_number":860029,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-05 14:40:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"recommended":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"all":[{"value":1,"date":"2022-10-04 14:29:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":2,"date":"2022-10-05 13:47:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2022-10-05 14:28:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2023-03-22 14:11:56.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"value":2,"date":"2022-10-05 13:58:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2022-10-03 11:46:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-05 14:19:59.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2022-10-05 14:28:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-30 19:39:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-03 08:29:24.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2022-10-04 14:29:14.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2022-10-05 13:47:58.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2022-10-05 13:58:35.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-10-05 14:19:59.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-03-22 14:11:56.000000000","updated_by":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"reviewer":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"state":"REVIEWER"}],"messages":[{"id":"a293a84c66047cffc7efcf447e40681411bf2e48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-30 19:25:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab738a004bca9acc4e7c129ea8ac236a7ec1b14e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-30 19:26:03.000000000","message":"Topic set to zed-final","accounts_in_message":[],"_revision_number":1},{"id":"a0aee4891243678e03cfcfc51a858b559c6636d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-30 19:27:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9f22a2d68e27f13b2f4d6a3d4134cadacf0098c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 19:39:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a15e44f740b54797b7bb559b05307f5c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13ada6909a5641b89652c99c9b324bbc : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/edb8b080257b4ec3a54ddfa49e71fe96 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb990fa158cd4cd785088f44cc3d3e1a : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c374daf3705945c5936178bd5c355150 : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f43f75c8448048879000794b5794f253 : SUCCESS in 8m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/601e362377e649bdb7ba7b2b49ba8a80 : SUCCESS in 5m 20s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8decb773e3d3445993c2e781ca8f7f70 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":2},{"id":"fabd002aeeb06423d98e3b058b70c56fe9fd4800","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-10-03 08:29:24.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3695fa015f32de009f82ce73b58264b2dd17f52d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-10-03 11:46:04.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"a4bd96a624b26f91b112863b704f7ec6816f5e95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 11:54:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbe430e2db44477ca115f5fe335d5f33\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7a6ed75f5ec4da58e0495802443a1b6 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d68c17219a9c411dbd7806065387e3d8 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/510e0d50dc154e8d93afd451de275bce : SUCCESS in 7m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/57ad08df931a497b82e640371615f61c : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c32dfeb7abd7462b9ef9d995a112f7c1 : SUCCESS in 6m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b8d1dbb516bc4b4baf3117d4f3cdf9c6 : SUCCESS in 5m 43s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/fbab5aab306a429cb76ee1e2905bb42c : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":3},{"id":"f8dede7bcb19077ac1401407d70119aa49cc93fd","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2022-10-04 14:29:14.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"624a4ff79a9ce47663250db9cb125b518d4acc56","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-10-05 13:47:58.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a3080ecd14accfd14d40c9c5c73fdbf01629496a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-10-05 13:58:35.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"86f401513ee1e77f7051b8e8be356a563ca425c2","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-10-05 14:19:59.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"dd13989df28ecc53734ab0fbfc00e92ab7a5a7f3","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-10-05 14:28:34.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a4cc75b8ec105a7076a5b0c08b16daed7f47d2b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 14:28:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e9f11e4eea2d81c26a7265116ee98af44fdaf9c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 14:40:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/497d5fb788dd4b3987c2fc2852e498e9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7119e7f6224c413391c99acacf7ced0c : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c895f2f2e00d436c84188275effade72 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bce9424fa3c24dc38d965dad09729924 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/badfc322a0144baca4468e8b7a08f133 : SUCCESS in 7m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ed12e5db8e9645a281b39bdee00ca918 : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82b952d406bf425bbe80b64ab44df29d : SUCCESS in 7m 30s","accounts_in_message":[],"_revision_number":3},{"id":"b959f467a4d8a8d39c657015d24b1f7a44d9180a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 14:40:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f86a3f2ce55288c04ee762d7e4f21872cf59779f","author":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"date":"2023-03-22 14:11:56.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"63433942e0a8c3c1c09daf2581ad8a0a0e4d0ed4","revisions":{"60e93b386700cd8dfdd29a57ebd04b7a27d52329":{"kind":"REWORK","_number":1,"created":"2022-09-30 19:25:44.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/29/860029/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/29/860029/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/860029/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/860029/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/860029/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/29/860029/1"}}},"commit":{"parents":[{"commit":"20822646c2b2a0800f411077e5454fefe4d4e994","subject":"Zed final releases for cycle-with-rc projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/20822646c2b2a0800f411077e5454fefe4d4e994"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-09-30 19:23:40.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-09-30 19:23:40.000000000","tz":120},"subject":"Mark Zed as released","message":"Mark Zed as released\n\nChange-Id: I65d51978ab978b66d6d67c5c323b29dc59c62de8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/60e93b386700cd8dfdd29a57ebd04b7a27d52329"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/60e93b386700cd8dfdd29a57ebd04b7a27d52329"}]},"branch":"refs/heads/master"},"04de2d6927ee0401b98469a9f9364968e99e65c4":{"kind":"REWORK","_number":2,"created":"2022-09-30 19:27:09.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/29/860029/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/29/860029/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/860029/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/860029/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/860029/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/29/860029/2"}}},"commit":{"parents":[{"commit":"20822646c2b2a0800f411077e5454fefe4d4e994","subject":"Zed final releases for cycle-with-rc projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/20822646c2b2a0800f411077e5454fefe4d4e994"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-09-30 19:23:40.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-09-30 19:26:35.000000000","tz":120},"subject":"Mark Zed as released","message":"Mark Zed as released\n\nChange-Id: I65d51978ab978b66d6d67c5c323b29dc59c62de8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/04de2d6927ee0401b98469a9f9364968e99e65c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/04de2d6927ee0401b98469a9f9364968e99e65c4"}]},"branch":"refs/heads/master"},"63433942e0a8c3c1c09daf2581ad8a0a0e4d0ed4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-10-03 11:46:04.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/29/860029/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/29/860029/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/860029/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/860029/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/860029/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/29/860029/3"}}},"commit":{"parents":[{"commit":"b38f01c4271fdb3919cb2ecafe94a72107f65f11","subject":"Zed final releases for cycle-with-rc projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b38f01c4271fdb3919cb2ecafe94a72107f65f11"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-09-30 19:23:40.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-10-03 11:46:04.000000000","tz":0},"subject":"Mark Zed as released","message":"Mark Zed as released\n\nChange-Id: I65d51978ab978b66d6d67c5c323b29dc59c62de8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/63433942e0a8c3c1c09daf2581ad8a0a0e4d0ed4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/63433942e0a8c3c1c09daf2581ad8a0a0e4d0ed4"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[]}
