)]}'
{"id":"openstack%2Freleases~861750","triplet_id":"openstack%2Freleases~master~I9d79b042a98fb5c5106e5183cee635274d179b3e","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-10-27 09:19:48.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I9d79b042a98fb5c5106e5183cee635274d179b3e","subject":"Move OpenStack-Ansible os_tacker and os_panko to EOL on pike","status":"ABANDONED","created":"2022-10-18 12:32:10.000000000","updated":"2022-10-27 09:19:48.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bca313d97d2b2d1187d8165131ed6d8119c578f6","_number":861750,"virtual_id_number":861750,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-10-18 12:39:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":-2,"date":"2022-10-18 12:43:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-10-18 12:33:05.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-18 12:33:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-18 12:43:56.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"f6cfa2ec8ad9d81641a569e3b490407889a9fd59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-18 12:32:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"481504d6e279a60f53613ec0a339ef56e7dc8b02","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 12:33:05.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"0d4471465addbb57ce281a97494f9e13b25666ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 12:39:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a2385f2cb254ba2ac9323b3ee1954d2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9878cad269954b78b048341bb0748130 : SUCCESS in 5m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6249b26493ec44b38af85bff8e1587c4 : SUCCESS in 4m 58s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/687271f4fe2c407caf1fa99a9d1937b6 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"af17a32f87085d6167d80915c2ac142f214f1af1","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-10-18 12:43:56.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bca313d97d2b2d1187d8165131ed6d8119c578f6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-10-27 09:19:48.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2d856a8f3d675e8ea251c67a4e038b1f95e3dc84","revisions":{"2d856a8f3d675e8ea251c67a4e038b1f95e3dc84":{"kind":"REWORK","_number":1,"created":"2022-10-18 12:32:10.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/50/861750/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/50/861750/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/861750/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/861750/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/861750/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/50/861750/1"}}},"commit":{"parents":[{"commit":"28bdf0cea8a331dafb248345117f8239b8c0b7db","subject":"Fix scheduled dates to Thursdays where needed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/28bdf0cea8a331dafb248345117f8239b8c0b7db"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-18 12:32:01.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-10-18 12:32:01.000000000","tz":120},"subject":"Move OpenStack-Ansible os_tacker and os_panko to EOL on pike","message":"Move OpenStack-Ansible os_tacker and os_panko to EOL on pike\n\nWhen EOLing most of repositories, os_tacker and os_panko were somehow\nmissed. We should EOL them and drop branches.\n\nChange-Id: I9d79b042a98fb5c5106e5183cee635274d179b3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2d856a8f3d675e8ea251c67a4e038b1f95e3dc84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2d856a8f3d675e8ea251c67a4e038b1f95e3dc84"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
