)]}'
{"id":"openstack%2Freleases~862341","triplet_id":"openstack%2Freleases~master~Id2a2c2dc969e55bc78a9ce60286de3044e01d8bd","project":"openstack/releases","branch":"master","topic":"wallaby-em","attention_set":{},"removed_from_attention_set":{"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2022-11-07 14:08:52.000000000","reason":"Change was submitted"},"8833":{"account":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"last_update":"2022-11-05 03:20:24.000000000","reason":"\u003cGERRIT_ACCOUNT_8833\u003e replied on the change","reason_account":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-11-07 14:08:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id2a2c2dc969e55bc78a9ce60286de3044e01d8bd","subject":"[tripleo] Transition Wallaby to EM","status":"MERGED","created":"2022-10-21 13:07:31.000000000","updated":"2022-11-07 14:08:52.000000000","submitted":"2022-11-07 14:08:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862341-wallaby-em","meta_rev_id":"2e0ffbdd2e051cb0186edf8ffabf1bb03b0ffd08","_number":862341,"virtual_id_number":862341,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-07 14:08:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"recommended":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":1,"date":"2022-11-05 03:20:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":2,"date":"2022-11-07 13:56:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2022-11-07 10:20:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":1,"date":"2022-11-07 13:56:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-11-05 03:21:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-21 13:08:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-21 13:19:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-21 13:20:13.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2022-10-21 13:20:14.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2022-11-07 10:20:36.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-11-07 13:56:29.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"e249383456245b6d1705f3a5a9fb3141de484d86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-10-21 13:07:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2efecae5b9e17d2e9511c63fa3993e5eaed4241","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 13:19:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9006bd896cc74803b405a4997a74412d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff8c37960c6a452aad320667d0d09c00 : SUCCESS in 7m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/36a43b2bd6d041299a85a785b8321a09 : FAILURE in 5m 56s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9900ed9680ce4c228fbe91629a162b27 : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":1},{"id":"d615af37919f808bbce110bcd77e027e5d648c2f","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-10-22 13:53:19.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e61dc4c450eee7dd2e3bf9c89945590049dd9552","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-10-24 07:01:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e15af5c8a89998fdf2b58b6b5d2939ce6044d0ff","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-10-24 09:43:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a44c5783d9ce039d0e616a1d6b286681a6f92841","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-11-04 14:38:52.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba5f257ca34c85862721814f8112cd7665a272ab","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 14:39:44.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"d11bcbbcd0239bb081fc127749094af889966701","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-11-04 16:50:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"09500abe60bcf0af01f19123919f2ec5026c409a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-11-04 17:05:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0d0a4060d629a1f89d27bfaf857d6e709f476ca8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 17:20:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91ad15e70b044d84a3bd64359e794956\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32b72d6d12ea4bbba52c165df3a5b985 : SUCCESS in 7m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/622a6a6932254171a30712902df6d6f5 : SUCCESS in 6m 23s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/67b3109e17ee47079eec20979bebdee8 : SUCCESS in 7m 50s","accounts_in_message":[],"_revision_number":2},{"id":"da34c39ff3dd97f62bd1182fd3e65aa6324aa468","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-11-05 03:20:24.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5fe265fa0a954e5c2a37d7364e4854eaa7b03f78","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-05 03:21:16.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"43abbfb37f3bdf24d2d191525e479dcb55f4611d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-07 10:20:36.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e730815d748ba5d37d0b1b61bf49e3902ad1923a","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-11-07 13:56:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5be93ed6245bd434c3e77de2c9b79444a5895d07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 13:57:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b748c5d3b548c0af49f4ef51f0b6771a3649ba0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 14:08:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/876791d336c0407aa899fd4df754df4e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37154413d29a4ef2839abf3b7f82d9a4 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a87a1677add47c0b9ece255fdcbda08 : SUCCESS in 5m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4abbfbc0d9cd464380100bd7855d9715 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":2},{"id":"2e0ffbdd2e051cb0186edf8ffabf1bb03b0ffd08","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 14:08:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a0faf09984c21d7632dad83bb87d4d4f0f5aee3a","revisions":{"2dedb98750055f1800e907741382efd510946bfc":{"kind":"REWORK","_number":1,"created":"2022-10-21 13:07:31.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/41/862341/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/41/862341/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/862341/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/862341/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/862341/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/41/862341/1"}}},"commit":{"parents":[{"commit":"802dfa1ab7f6c891aeba0b55128680836ca4b0d8","subject":"Merge \"EOL Stein for all Ironic-program projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/802dfa1ab7f6c891aeba0b55128680836ca4b0d8"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-10-21 12:35:13.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-10-21 12:35:13.000000000","tz":120},"subject":"[tripleo] Transition Wallaby to EM","message":"[tripleo] Transition Wallaby to EM\n\nThis transition the wallaby branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nThe transition deadline is November 2nd, 2022.\n\nChange-Id: Id2a2c2dc969e55bc78a9ce60286de3044e01d8bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2dedb98750055f1800e907741382efd510946bfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2dedb98750055f1800e907741382efd510946bfc"}]},"branch":"refs/heads/master"},"a0faf09984c21d7632dad83bb87d4d4f0f5aee3a":{"kind":"REWORK","_number":2,"created":"2022-11-04 16:50:37.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/41/862341/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/41/862341/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/862341/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/862341/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/41/862341/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/41/862341/2"}}},"commit":{"parents":[{"commit":"802dfa1ab7f6c891aeba0b55128680836ca4b0d8","subject":"Merge \"EOL Stein for all Ironic-program projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/802dfa1ab7f6c891aeba0b55128680836ca4b0d8"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-10-21 12:35:13.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-11-04 16:49:51.000000000","tz":60},"subject":"[tripleo] Transition Wallaby to EM","message":"[tripleo] Transition Wallaby to EM\n\nThis transition the wallaby branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nThe transition deadline is November 2nd, 2022.\n\nChange-Id: Id2a2c2dc969e55bc78a9ce60286de3044e01d8bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a0faf09984c21d7632dad83bb87d4d4f0f5aee3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a0faf09984c21d7632dad83bb87d4d4f0f5aee3a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[]}
