)]}'
{"id":"openstack%2Freleases~862672","triplet_id":"openstack%2Freleases~master~Ie7af9ff626cabfb75fadff6860123db2a38a4710","project":"openstack/releases","branch":"master","topic":"validations-libs/1.7.2","attention_set":{},"removed_from_attention_set":{"32926":{"account":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"last_update":"2022-10-26 11:14:21.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie7af9ff626cabfb75fadff6860123db2a38a4710","subject":"Releasing validations-libs 1.7.2","status":"ABANDONED","created":"2022-10-26 08:09:11.000000000","updated":"2022-10-26 11:14:21.000000000","total_comment_count":4,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"d7169d8a766ebe5764352eccc0143961381b3e1f","_number":862672,"virtual_id_number":862672,"owner":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-10-26 10:10:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-26 08:09:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-26 08:21:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-26 08:52:14.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"1e85947ea868a79b5525de750b58da41b8f673bb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-10-26 08:09:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a73ad25906f7946420cbb7d77f1d8a06089f723e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-26 08:21:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5ca582d3705413eb1afce1f61d857a4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8fd96064260428d99d5af87356c2b66 : SUCCESS in 7m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5cf03ad5742e4401a3d7866126207d4f : SUCCESS in 5m 24s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/09453081cb144d0eb8200588fd9fc368 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"27985c4b747497dfbed89147e9e4092d9c32cc05","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-10-26 08:28:12.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"c809431653cf41c78e87419bd65c0604c51bf2df","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-10-26 08:52:14.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"dcf80ab43303d1bada7ac930b19ea5661e5928a5","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-10-26 09:13:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d27fed8da046b81f83e1adb4d3977d312d3870ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-10-26 09:30:41.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"a6d48d532aa11859e8c581bd0d3bf864ceb35764","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-26 09:39:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8468f3e7d9394d42a9940bdbd5ad87ee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53b8e6808244470596bdccee50ec2e2d : SUCCESS in 7m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/55bc3c75a39543c385f6c14253bfcdd6 : FAILURE in 5m 45s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7c3520b9162d4cc7874f9d9be525a44f : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":2},{"id":"72cc698cf14afe6e7e06583f3d39c6e9c9ca108d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-10-26 10:04:16.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"d04d6022f661a5b748fa78971b1faaefc02ad312","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-26 10:10:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d07867bc9db4afebda008aa8007be7d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9ed7dfc2de34ff1878ba1ee6ec21548 : SUCCESS in 5m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cf48e917bd504691aff32fa0fee34d3d : SUCCESS in 4m 32s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/af6b46c145f94a809a54c429d8c2f1a5 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":3},{"id":"d7169d8a766ebe5764352eccc0143961381b3e1f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-10-26 11:14:21.000000000","message":"Abandoned\n\nMajor release needed","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a380bc7cf0bc7d21b03df15f86ef0d5e85395fdb","revisions":{"2be8871de2e69003637cc4b8f2bdb1a3fe3d54f1":{"kind":"REWORK","_number":1,"created":"2022-10-26 08:09:11.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/72/862672/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/72/862672/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/862672/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/862672/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/862672/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/72/862672/1"}}},"commit":{"parents":[{"commit":"e5a800b5711d29e1ab15c961027092f9c11a3d8c","subject":"Merge \"Release ironic-python-agent for stable/wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e5a800b5711d29e1ab15c961027092f9c11a3d8c"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-10-26 07:53:11.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-10-26 08:07:00.000000000","tz":120},"subject":"Releasing validations-libs 1.7.2","message":"Releasing validations-libs 1.7.2\n\nRelease contains parser fixes, allowing for expanded dictionary functionality,\nremoval of unused legacy API ported from tripleoclient, and stylistic adjustments.\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: Ie7af9ff626cabfb75fadff6860123db2a38a4710\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2be8871de2e69003637cc4b8f2bdb1a3fe3d54f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2be8871de2e69003637cc4b8f2bdb1a3fe3d54f1"}]},"branch":"refs/heads/master"},"34c25250ba468156f97db2a0fd7b84a658d977ff":{"kind":"REWORK","_number":2,"created":"2022-10-26 09:30:41.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/72/862672/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/72/862672/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/862672/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/862672/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/862672/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/72/862672/2"}}},"commit":{"parents":[{"commit":"e5a800b5711d29e1ab15c961027092f9c11a3d8c","subject":"Merge \"Release ironic-python-agent for stable/wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e5a800b5711d29e1ab15c961027092f9c11a3d8c"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-10-26 07:53:11.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-10-26 09:30:41.000000000","tz":0},"subject":"Releasing validations-libs 1.7.2","message":"Releasing validations-libs 1.7.2\n\nRelease contains parser fixes, allowing for expanded dictionary functionality.\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: Ie7af9ff626cabfb75fadff6860123db2a38a4710\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/34c25250ba468156f97db2a0fd7b84a658d977ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/34c25250ba468156f97db2a0fd7b84a658d977ff"}]},"branch":"refs/heads/master"},"a380bc7cf0bc7d21b03df15f86ef0d5e85395fdb":{"kind":"REWORK","_number":3,"created":"2022-10-26 10:04:16.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/72/862672/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/72/862672/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/862672/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/862672/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/862672/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/72/862672/3"}}},"commit":{"parents":[{"commit":"e5a800b5711d29e1ab15c961027092f9c11a3d8c","subject":"Merge \"Release ironic-python-agent for stable/wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e5a800b5711d29e1ab15c961027092f9c11a3d8c"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-10-26 07:53:11.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-10-26 10:04:12.000000000","tz":0},"subject":"Releasing validations-libs 1.7.2","message":"Releasing validations-libs 1.7.2\n\nRelease contains parser fixes, allowing for expanded dictionary functionality.\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: Ie7af9ff626cabfb75fadff6860123db2a38a4710\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a380bc7cf0bc7d21b03df15f86ef0d5e85395fdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a380bc7cf0bc7d21b03df15f86ef0d5e85395fdb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
