)]}'
{"id":"openstack%2Freleases~862674","triplet_id":"openstack%2Freleases~master~I3b2f9e3b970cd8273712df2d84f79ebb18701377","project":"openstack/releases","branch":"master","topic":"validations-libs/1.7.4","hashtags":[],"change_id":"I3b2f9e3b970cd8273712df2d84f79ebb18701377","subject":"Releasing validations-libs 1.7.4","status":"ABANDONED","created":"2022-10-26 08:25:28.000000000","updated":"2022-10-26 11:14:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"118cde0f387a81d417ba35f4504a08fb00daf4b9","_number":862674,"virtual_id_number":862674,"owner":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-10-26 10:04:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2022-10-26 08:26:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-26 08:51:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b97e50e6f80dd38e574bad5bbeafeecefe7df49","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-10-26 08:25:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"745a48a192d4c6a62e207a0c5836ee0fbe0c6622","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-26 08:51:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c574a85a683b4e8eaebad6721d51733e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8d5708817e24f1f9c6f08df8064539c : SUCCESS in 5m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a6bc13be4fde4d25b3d90365e7a3a24b : FAILURE in 5m 26s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4c84983bf50f4f11b10e1a45aa815b2d : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"5af6df92c7f64a12809fa8636312eeb2ccf15012","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-10-26 09:50:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"535813cb30af4ea4f0c44e143750dcb9718d435a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-26 10:04:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d15b98d3b01c474598cdf08bd622b343\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c4e9c062a0747ee8cfe0e2df1c9a3ad : SUCCESS in 6m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/013146f17f1d4092b84a02271126714d : FAILURE in 5m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/386a88fab07c433684aa073b0d19771b : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":2},{"id":"118cde0f387a81d417ba35f4504a08fb00daf4b9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-10-26 11:14:32.000000000","message":"Abandoned\n\nMajor release needed","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9f20e4900caecad809fc99ef2558ccf6d4bc0717","revisions":{"b16fce079febfa59844e1daa91a944b42b0f66a2":{"kind":"REWORK","_number":1,"created":"2022-10-26 08:25:28.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/74/862674/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/74/862674/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/862674/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/862674/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/862674/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/74/862674/1"}}},"commit":{"parents":[{"commit":"5ad66319e30cb3845cbbbe6d4b9a50b8e4c134e4","subject":"Releasing validations-libs 1.7.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5ad66319e30cb3845cbbbe6d4b9a50b8e4c134e4"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-10-26 08:22:34.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-10-26 08:24:46.000000000","tz":120},"subject":"Releasing validations-libs 1.7.4","message":"Releasing validations-libs 1.7.4\n\nRelease includes fixes to validations runtime logging facilities.\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I3b2f9e3b970cd8273712df2d84f79ebb18701377\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b16fce079febfa59844e1daa91a944b42b0f66a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b16fce079febfa59844e1daa91a944b42b0f66a2"}]},"branch":"refs/heads/master"},"9f20e4900caecad809fc99ef2558ccf6d4bc0717":{"kind":"REWORK","_number":2,"created":"2022-10-26 09:50:54.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/74/862674/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/74/862674/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/862674/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/862674/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/862674/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/74/862674/2"}}},"commit":{"parents":[{"commit":"c28f2d09c41cc5ed170102d26529bfd63b57cae1","subject":"Releasing validations-libs 1.7.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c28f2d09c41cc5ed170102d26529bfd63b57cae1"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-10-26 08:22:34.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-10-26 09:50:06.000000000","tz":120},"subject":"Releasing validations-libs 1.7.4","message":"Releasing validations-libs 1.7.4\n\nRelease includes fixes to validations runtime logging facilities.\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I3b2f9e3b970cd8273712df2d84f79ebb18701377\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9f20e4900caecad809fc99ef2558ccf6d4bc0717"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9f20e4900caecad809fc99ef2558ccf6d4bc0717"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
