)]}'
{"id":"openstack%2Freleases~863011","triplet_id":"openstack%2Freleases~master~Id6fa1b598ee430a6b496cf4f315e5fb3e7a2df4a","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"8833":{"account":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"last_update":"2022-11-02 16:10:06.000000000","reason":"Change was submitted"},"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2022-11-02 07:39:06.000000000","reason":"\u003cGERRIT_ACCOUNT_8449\u003e replied on the change","reason_account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}},"16515":{"account":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"last_update":"2022-11-02 08:17:56.000000000","reason":"\u003cGERRIT_ACCOUNT_16515\u003e replied on the change","reason_account":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"}},"27427":{"account":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"last_update":"2022-10-31 14:08:34.000000000","reason":"\u003cGERRIT_ACCOUNT_27427\u003e replied on the change","reason_account":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-11-02 08:31:32.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-11-02 16:10:06.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-11-02 15:56:16.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},"32926":{"account":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"last_update":"2022-11-02 07:48:55.000000000","reason":"\u003cGERRIT_ACCOUNT_32926\u003e replied on the change","reason_account":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}}},"hashtags":[],"change_id":"Id6fa1b598ee430a6b496cf4f315e5fb3e7a2df4a","subject":"Create stable/zed for tripleo repos","status":"MERGED","created":"2022-10-31 06:05:25.000000000","updated":"2022-11-02 16:10:06.000000000","submitted":"2022-11-02 16:10:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":23,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863011","meta_rev_id":"4b90992ee37a121ca4a730d412b1e968938042b1","_number":863011,"virtual_id_number":863011,"owner":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-02 16:10:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"all":[{"value":1,"date":"2022-11-02 07:39:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":1,"date":"2022-11-02 08:17:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":2,"date":"2022-11-02 08:31:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2022-11-02 05:47:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2022-11-02 10:56:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":2,"date":"2022-11-02 15:56:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-02 07:48:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":1,"date":"2022-11-02 15:56:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-11-02 05:27:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"CC":[{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-31 06:06:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-31 06:09:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-31 06:30:10.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2022-10-31 06:31:19.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2022-10-31 06:31:19.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"state":"REVIEWER"},{"updated":"2022-10-31 07:16:56.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2022-10-31 07:16:56.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"state":"REVIEWER"},{"updated":"2022-10-31 07:16:56.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"state":"REVIEWER"},{"updated":"2022-10-31 07:19:00.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2022-11-02 06:00:24.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-11-02 06:00:24.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2022-11-02 08:40:06.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"CC"},{"updated":"2022-11-02 10:56:57.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2022-11-02 14:09:44.000000000","updated_by":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"reviewer":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"state":"CC"}],"messages":[{"id":"e789030078f16c58aea0dae561f7f4bf5b626ccf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-10-31 06:05:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a752f148d7e0eb020ab85c8254f3691708457a60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 06:09:33.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7599e3b81ab4324812d3fbab4045283\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b0283493cb3417ab5a70c7503f11bce : FAILURE in 3m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9aba6b80764341e0a42a44ef808e1a4f : FAILURE in 3m 00s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/478ab0fab41b4cc58c66605ec72d1bd9 : FAILURE in 3m 07s","accounts_in_message":[],"_revision_number":1},{"id":"969e65e1ffb3e7a6c9e964d8b07b8ec1b452f94d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-10-31 06:28:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"84ca6cb2a9f7d8facf1a0b7e9a7a88ae0595f6be","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 06:29:30.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"eb8ec2c4fabe1b04943b81df8ce09383f555a646","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-10-31 06:29:58.000000000","message":"Removed PTL-Approved+1 by \u003cGERRIT_ACCOUNT_22348\u003e\n","accounts_in_message":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"_revision_number":2},{"id":"ef88e7a951012a0bd9f51ca2826a86eacf6b9710","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-31 06:30:10.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0013887a718faee77d64e74473c9526ef9e2370a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-31 06:31:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1552843811d01a0da33fef4be56cfe7c20223dd6","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 06:31:51.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"688bfad8b8a68237b7ad70a23133b178d212cf62","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-10-31 06:41:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"46a172934826c73e4c7c49ed0c5136fd85a35f90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 06:42:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/59f77ed73ee74cf5a1ac23e4135ba016\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d5892327211405cb001c8f4ade7f0ec : SUCCESS in 6m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/af272505433640c2a4f4a1002102394c : FAILURE in 9m 02s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/52444ad481664f4fa6fb41d6a4fee126 : FAILURE in 4m 33s","accounts_in_message":[],"_revision_number":2},{"id":"b9804cd34038f1e26a054c04082f2f25da3e38cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-10-31 06:51:40.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"36cd36705433ed1dae6a7fa60cce4f4187af9e70","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 06:53:28.000000000","message":"Patch Set 3: Verified-1 PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"4ba9e8f15daffc94d339f1f639dcc9e22db4b7e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-10-31 06:56:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"185ca978ee0a7059dd539f0b2378c7b5ceeee0e5","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 06:57:37.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"099e8a4de37789e66b087652750e736c91f5933e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 07:15:10.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac3a4b39afe64743b0edaf92b3d92c35\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/755ab9cd8c064dd0ae0e043ad76cc4b5 : SUCCESS in 6m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/92f3b71362054c11a633018ee9713c37 : FAILURE in 13m 42s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3bcbfcd612b84af9bb88e1d597b3a06b : FAILURE in 4m 27s","accounts_in_message":[],"_revision_number":4},{"id":"80b00460c3aef15914e329a0ff7e14c61bfa21a0","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-10-31 07:16:56.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"5ea06e6fb849d15382784ccac4b9c93648590a0d","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-10-31 07:26:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e1ede3c2c8689a7fecd7bef9890e36f7652af476","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-10-31 07:45:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c91d7bd8aa5fa42a013809cc39ca95379c1bf178","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-10-31 08:16:36.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8969b27f9960d91eec5288ef3adc2045d18402fe","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 08:17:35.000000000","message":"Patch Set 5: PTL-Approved+1","accounts_in_message":[],"_revision_number":5},{"id":"669fbd2e109e049a6b752ea4fb75186fe8c032b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 08:36:42.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/36c2a0a511274291aa4d42c0bb1b32bc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aa5b5bc45814d69bf294595fdc5faba : SUCCESS in 6m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/390bbcbe43d747d59e02f41ca13d30c6 : FAILURE in 9m 14s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/34012e9c1a414f7286d5b309112f49d7 : FAILURE in 4m 58s","accounts_in_message":[],"_revision_number":5},{"id":"5d052a7fdaf6b9916328b2ef9ab6e94f93c5fb40","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-10-31 13:50:45.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"6ccd7db93051f875267321779aa6b8cf59e5921e","author":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"date":"2022-10-31 14:08:34.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d4a12cd65bcfd23d1a253afcffd5ddcb2913e1d0","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-10-31 14:20:57.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"20a8a4e58dd67fa3aab3df248376f414bf0f75d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-10-31 18:38:47.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c1cf8c765cef51c8e81bf418fc9196a5e575c6fe","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 18:39:34.000000000","message":"Patch Set 6: PTL-Approved+1","accounts_in_message":[],"_revision_number":6},{"id":"96e36b05c0f1dde0f7f4ae225209de16f2702878","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-10-31 18:41:54.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"853e67b4e09f64b0dbb0ad121eea90a37c421c85","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 18:42:43.000000000","message":"Patch Set 7: PTL-Approved+1","accounts_in_message":[],"_revision_number":7},{"id":"853a7ba7cc20461e0bd04be7fe0d9dc47435023b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-10-31 18:52:30.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"1cd27dcb561f565090287b9a0776547b61651d00","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 18:53:21.000000000","message":"Patch Set 8: PTL-Approved+1","accounts_in_message":[],"_revision_number":8},{"id":"6f25189cd2754267cc5b5bed1f1502fb8c9471dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 19:03:28.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4cae3f21b034d669af73f4ff32943c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8e78c391d314bcea86fbc05a774637f : SUCCESS in 5m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/051da83672934b30ade3e81fb275f72c : FAILURE in 9m 51s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/04502256042d49a890ff4aa4de3ad39d : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":8},{"id":"3fc74228983d90c923e7b1f4c22cccc344d27ea6","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-10-31 19:28:09.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e8c9717ad8e1e19f5fb9f0047f15a6b9d8d9339d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-11-02 04:43:25.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"7b834282ee7a443d83d49a4acee0ff16aa7c848c","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 04:44:16.000000000","message":"Patch Set 9: PTL-Approved+1","accounts_in_message":[],"_revision_number":9},{"id":"e040965e24525a563ee90f0451e05eb804682683","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-02 04:47:36.000000000","message":"Patch Set 9: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"a1397068eae7fff497bb8cda1cf36521ba938025","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 04:57:13.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8841326b8a504ad38a55b4b6531e501a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69ade8c404f44204a18bb7b8705f7efe : SUCCESS in 5m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2c19d1dfb836437d9ee6b556a129175d : FAILURE in 10m 13s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1b4163de4ba248cca27b2e69be118b70 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":9},{"id":"0f0857dde7c8a500cdba466b24d229dd84dceec9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-11-02 05:14:00.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"c15917df961a35441c8da819752d74576e33904e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 05:14:47.000000000","message":"Patch Set 10: PTL-Approved+1","accounts_in_message":[],"_revision_number":10},{"id":"ebe98320a8a6acf658d40e5e8bbf86bb3fbc75b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 05:25:46.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0f62f82bdc14b38b93e162e82eb0dc8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc3ae6f9816043c383290ee973e6278d : SUCCESS in 5m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3b21bbab8e084e06b0426effb18b1e7e : FAILURE in 10m 17s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6b88415004cb43b49ac02f3bc8b4fbbd : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":10},{"id":"bcd51726334a836feee15fcb94d1482469f5fd83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-11-02 05:26:42.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"e65528143bad89734302a5b7fc24c5987064abf6","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 05:27:41.000000000","message":"Patch Set 11: PTL-Approved+1","accounts_in_message":[],"_revision_number":11},{"id":"d702e2efb98e74f0fb6ca96913635022cb6d47fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 05:42:37.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9a0e585ad704eb99d6253d96c64db62\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfad619eb2db4804bd74789241d1b70f : SUCCESS in 6m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5c53688ffb354062bb22374e1cc28d22 : SUCCESS in 12m 50s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/04f7f3cfdc2d4f05b3ef99f12f4d7a39 : SUCCESS in 8m 04s","accounts_in_message":[],"_revision_number":11},{"id":"e1e1eb21955b0d70a71172166bf9a199904d226d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-02 05:47:13.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"28e25722857e87952b0117ac2e08da737608791a","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-11-02 05:52:29.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"ffabdc1468fbc5da4cd01da27bdff458e06b8676","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-11-02 05:53:32.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"1dbecee6ed122e9d545ca9fd8e9504443793a5c5","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-11-02 07:39:06.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"10b18bf59efe05b17afd6cd40ccec2bc9956cf78","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-11-02 07:48:55.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"71133b0e51d3208245589f76d14a96564c0f5a19","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2022-11-02 08:17:56.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"d03fd54559649a9fc0546cec49baf2e2ee44ad30","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-11-02 08:31:32.000000000","message":"Patch Set 11: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"fdbce5f9173330dd80d863a71a1fef1e5dc73d22","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-11-02 10:56:57.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"83604c3440dd620792c9cbb5abdadb4b440df8f0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-02 15:56:16.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"6a6827c7d3a6b2dcc49829f26a82ca9ac1ecf3f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 15:56:40.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"ac9834a0573781189c76f3faeb523382a0aab681","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 16:10:06.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8771b386b5f943b0b445b9ec76cca738\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1682d2f1e95e491ea9bd65d2edbdd5b4 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b296a677975549cbb69c66e4e91de434 : SUCCESS in 6m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3c29e7149318486992db40268cc52c83 : SUCCESS in 12m 01s","accounts_in_message":[],"_revision_number":11},{"id":"4b90992ee37a121ca4a730d412b1e968938042b1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 16:10:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"d53c5e3e96588355fe5f8635ea31e4157b82bee2","revisions":{"d927ec14faa7df99ccd8a2a24359da1542db5761":{"kind":"REWORK","_number":1,"created":"2022-10-31 06:05:25.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/11/863011/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/11/863011/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/11/863011/1"}}},"commit":{"parents":[{"commit":"d6031738ef67a6d7741f353413e1dd63b28e6f4c","subject":"Merge \"[zaqar] Transition Wallaby to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6031738ef67a6d7741f353413e1dd63b28e6f4c"}]}],"author":{"name":"rabi","email":"ramishra@redhat.com","date":"2022-10-31 05:58:49.000000000","tz":330},"committer":{"name":"rabi","email":"ramishra@redhat.com","date":"2022-10-31 06:05:21.000000000","tz":330},"subject":"WIP Release tripleo zed","message":"WIP Release tripleo zed\n\nDepends-on: https://review.opendev.org/c/openstack/puppet-tripleo/+/863010\nDepends-On: https://review.opendev.org/c/openstack/puppet-pacemaker/+/862989\nChange-Id: Id6fa1b598ee430a6b496cf4f315e5fb3e7a2df4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d927ec14faa7df99ccd8a2a24359da1542db5761"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d927ec14faa7df99ccd8a2a24359da1542db5761"}]},"branch":"refs/heads/master"},"4c3667fbab95b0f7d4ad205b85ad77e4e6cc4db9":{"kind":"REWORK","_number":2,"created":"2022-10-31 06:28:35.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/11/863011/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/11/863011/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/11/863011/2"}}},"commit":{"parents":[{"commit":"d6031738ef67a6d7741f353413e1dd63b28e6f4c","subject":"Merge \"[zaqar] Transition Wallaby to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6031738ef67a6d7741f353413e1dd63b28e6f4c"}]}],"author":{"name":"rabi","email":"ramishra@redhat.com","date":"2022-10-31 05:58:49.000000000","tz":330},"committer":{"name":"rabi","email":"ramishra@redhat.com","date":"2022-10-31 06:27:52.000000000","tz":330},"subject":"WIP Release tripleo zed","message":"WIP Release tripleo zed\n\nDepends-on: https://review.opendev.org/c/openstack/puppet-tripleo/+/863010\nDepends-On: https://review.opendev.org/c/openstack/puppet-pacemaker/+/862989\nChange-Id: Id6fa1b598ee430a6b496cf4f315e5fb3e7a2df4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4c3667fbab95b0f7d4ad205b85ad77e4e6cc4db9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4c3667fbab95b0f7d4ad205b85ad77e4e6cc4db9"}]},"branch":"refs/heads/master"},"7177fa8beb687a80c707eff8e0d40e2565d2db75":{"kind":"NO_CHANGE","_number":3,"created":"2022-10-31 06:51:40.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/11/863011/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/11/863011/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/11/863011/3"}}},"commit":{"parents":[{"commit":"d6031738ef67a6d7741f353413e1dd63b28e6f4c","subject":"Merge \"[zaqar] Transition Wallaby to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6031738ef67a6d7741f353413e1dd63b28e6f4c"}]}],"author":{"name":"rabi","email":"ramishra@redhat.com","date":"2022-10-31 05:58:49.000000000","tz":330},"committer":{"name":"rabi","email":"ramishra@redhat.com","date":"2022-10-31 06:51:27.000000000","tz":330},"subject":"WIP Release tripleo zed","message":"WIP Release tripleo zed\n\nDepends-on: https://review.opendev.org/c/openstack/puppet-tripleo/+/863010\nDepends-On: https://review.opendev.org/c/openstack/puppet-pacemaker/+/862989\nChange-Id: Id6fa1b598ee430a6b496cf4f315e5fb3e7a2df4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7177fa8beb687a80c707eff8e0d40e2565d2db75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7177fa8beb687a80c707eff8e0d40e2565d2db75"}]},"branch":"refs/heads/master"},"fd9505acfcfd13a702475eba31be2f131ebff035":{"kind":"REWORK","_number":4,"created":"2022-10-31 06:56:35.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/11/863011/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/11/863011/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/11/863011/4"}}},"commit":{"parents":[{"commit":"d6031738ef67a6d7741f353413e1dd63b28e6f4c","subject":"Merge \"[zaqar] Transition Wallaby to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6031738ef67a6d7741f353413e1dd63b28e6f4c"}]}],"author":{"name":"rabi","email":"ramishra@redhat.com","date":"2022-10-31 05:58:49.000000000","tz":330},"committer":{"name":"rabi","email":"ramishra@redhat.com","date":"2022-10-31 06:55:54.000000000","tz":330},"subject":"WIP Release tripleo zed","message":"WIP Release tripleo zed\n\nDepends-on: https://review.opendev.org/c/openstack/puppet-tripleo/+/863010\nDepends-On: https://review.opendev.org/c/openstack/puppet-pacemaker/+/862989\nChange-Id: Id6fa1b598ee430a6b496cf4f315e5fb3e7a2df4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fd9505acfcfd13a702475eba31be2f131ebff035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fd9505acfcfd13a702475eba31be2f131ebff035"}]},"branch":"refs/heads/master"},"e603a0ae5cc4014ffeedc304d87ed70d8e27b523":{"kind":"REWORK","_number":5,"created":"2022-10-31 08:16:36.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/11/863011/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/11/863011/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/11/863011/5"}}},"commit":{"parents":[{"commit":"d6031738ef67a6d7741f353413e1dd63b28e6f4c","subject":"Merge \"[zaqar] Transition Wallaby to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6031738ef67a6d7741f353413e1dd63b28e6f4c"}]}],"author":{"name":"rabi","email":"ramishra@redhat.com","date":"2022-10-31 05:58:49.000000000","tz":330},"committer":{"name":"rabi","email":"ramishra@redhat.com","date":"2022-10-31 08:16:04.000000000","tz":330},"subject":"WIP Release tripleo zed","message":"WIP Release tripleo zed\n\nDepends-on: https://review.opendev.org/c/openstack/puppet-tripleo/+/863010\nDepends-On: https://review.opendev.org/c/openstack/puppet-pacemaker/+/862989\nChange-Id: Id6fa1b598ee430a6b496cf4f315e5fb3e7a2df4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e603a0ae5cc4014ffeedc304d87ed70d8e27b523"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e603a0ae5cc4014ffeedc304d87ed70d8e27b523"}]},"branch":"refs/heads/master"},"555028addf3b7fdc570f4decdc10ff4bbb64798a":{"kind":"REWORK","_number":6,"created":"2022-10-31 18:38:47.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/11/863011/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/11/863011/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/11/863011/6"}}},"commit":{"parents":[{"commit":"d6031738ef67a6d7741f353413e1dd63b28e6f4c","subject":"Merge \"[zaqar] Transition Wallaby to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6031738ef67a6d7741f353413e1dd63b28e6f4c"}]}],"author":{"name":"rabi","email":"ramishra@redhat.com","date":"2022-10-31 05:58:49.000000000","tz":330},"committer":{"name":"rabi","email":"ramishra@redhat.com","date":"2022-10-31 18:38:07.000000000","tz":330},"subject":"Create stable/zed for tripleo repos","message":"Create stable/zed for tripleo repos\n\nAfter moving to independent release model,\nwe did releases in Xena and Yoga without\nstable branches. For Zed we\u0027re cutting\nstable/zed branch along with releases\nfor all relevant TripleO repos.\n\nChange-Id: Id6fa1b598ee430a6b496cf4f315e5fb3e7a2df4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/555028addf3b7fdc570f4decdc10ff4bbb64798a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/555028addf3b7fdc570f4decdc10ff4bbb64798a"}]},"branch":"refs/heads/master"},"056f6dcd9e34a2a89ad080ec98fb901969ecd993":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2022-10-31 18:41:54.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/11/863011/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/11/863011/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/11/863011/7"}}},"commit":{"parents":[{"commit":"d6031738ef67a6d7741f353413e1dd63b28e6f4c","subject":"Merge \"[zaqar] Transition Wallaby to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6031738ef67a6d7741f353413e1dd63b28e6f4c"}]}],"author":{"name":"rabi","email":"ramishra@redhat.com","date":"2022-10-31 05:58:49.000000000","tz":330},"committer":{"name":"rabi","email":"ramishra@redhat.com","date":"2022-10-31 18:41:41.000000000","tz":330},"subject":"Create stable/zed for tripleo repos","message":"Create stable/zed for tripleo repos\n\nAfter moving to independent release model, we did releases\nfor Xena and Yoga without stable branches. For Zed we\u0027re\ncreating stable/zed branch along with releases for all\nrelevant TripleO repos.\n\nChange-Id: Id6fa1b598ee430a6b496cf4f315e5fb3e7a2df4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/056f6dcd9e34a2a89ad080ec98fb901969ecd993"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/056f6dcd9e34a2a89ad080ec98fb901969ecd993"}]},"branch":"refs/heads/master"},"8ecd8279a22bfb74125fc93c16201f9d515e5c99":{"kind":"REWORK","_number":8,"created":"2022-10-31 18:52:30.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/11/863011/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/11/863011/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/11/863011/8"}}},"commit":{"parents":[{"commit":"d6031738ef67a6d7741f353413e1dd63b28e6f4c","subject":"Merge \"[zaqar] Transition Wallaby to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6031738ef67a6d7741f353413e1dd63b28e6f4c"}]}],"author":{"name":"rabi","email":"ramishra@redhat.com","date":"2022-10-31 05:58:49.000000000","tz":330},"committer":{"name":"rabi","email":"ramishra@redhat.com","date":"2022-10-31 18:52:01.000000000","tz":330},"subject":"Create stable/zed for tripleo repos","message":"Create stable/zed for tripleo repos\n\nAfter moving to independent release model, we did releases\nfor Xena and Yoga without stable branches. For Zed we\u0027re\ncreating stable/zed branch along with releases for all\nrelevant TripleO repos.\n\nChange-Id: Id6fa1b598ee430a6b496cf4f315e5fb3e7a2df4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8ecd8279a22bfb74125fc93c16201f9d515e5c99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8ecd8279a22bfb74125fc93c16201f9d515e5c99"}]},"branch":"refs/heads/master"},"9a80fab464cf91f7009c6195496c4c6c9c57c795":{"kind":"REWORK","_number":9,"created":"2022-11-02 04:43:25.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/11/863011/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/11/863011/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/11/863011/9"}}},"commit":{"parents":[{"commit":"d6031738ef67a6d7741f353413e1dd63b28e6f4c","subject":"Merge \"[zaqar] Transition Wallaby to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6031738ef67a6d7741f353413e1dd63b28e6f4c"}]}],"author":{"name":"rabi","email":"ramishra@redhat.com","date":"2022-10-31 05:58:49.000000000","tz":330},"committer":{"name":"rabi","email":"ramishra@redhat.com","date":"2022-11-02 04:42:42.000000000","tz":330},"subject":"Create stable/zed for tripleo repos","message":"Create stable/zed for tripleo repos\n\nAfter moving to independent release model, we did releases\nfor Xena and Yoga without stable branches. For Zed we\u0027re\ncreating stable/zed branch along with releases for all\nrelevant TripleO repos.\n\nChange-Id: Id6fa1b598ee430a6b496cf4f315e5fb3e7a2df4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9a80fab464cf91f7009c6195496c4c6c9c57c795"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9a80fab464cf91f7009c6195496c4c6c9c57c795"}]},"branch":"refs/heads/master"},"8a8b8c82bb72bb8878fded1a4eec521d7e15157f":{"kind":"REWORK","_number":10,"created":"2022-11-02 05:14:00.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/11/863011/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/11/863011/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/11/863011/10"}}},"commit":{"parents":[{"commit":"d6031738ef67a6d7741f353413e1dd63b28e6f4c","subject":"Merge \"[zaqar] Transition Wallaby to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6031738ef67a6d7741f353413e1dd63b28e6f4c"}]}],"author":{"name":"rabi","email":"ramishra@redhat.com","date":"2022-10-31 05:58:49.000000000","tz":330},"committer":{"name":"rabi","email":"ramishra@redhat.com","date":"2022-11-02 05:13:49.000000000","tz":330},"subject":"Create stable/zed for tripleo repos","message":"Create stable/zed for tripleo repos\n\nAfter moving to independent release model, we did releases\nfor Xena and Yoga without stable branches. For Zed we\u0027re\ncreating stable/zed branch along with releases for all\nrelevant TripleO repos.\n\nChange-Id: Id6fa1b598ee430a6b496cf4f315e5fb3e7a2df4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8a8b8c82bb72bb8878fded1a4eec521d7e15157f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8a8b8c82bb72bb8878fded1a4eec521d7e15157f"}]},"branch":"refs/heads/master"},"d53c5e3e96588355fe5f8635ea31e4157b82bee2":{"kind":"REWORK","_number":11,"created":"2022-11-02 05:26:42.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/11/863011/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/11/863011/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/11/863011/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/11/863011/11"}}},"commit":{"parents":[{"commit":"d6031738ef67a6d7741f353413e1dd63b28e6f4c","subject":"Merge \"[zaqar] Transition Wallaby to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6031738ef67a6d7741f353413e1dd63b28e6f4c"}]}],"author":{"name":"rabi","email":"ramishra@redhat.com","date":"2022-10-31 05:58:49.000000000","tz":330},"committer":{"name":"rabi","email":"ramishra@redhat.com","date":"2022-11-02 05:26:29.000000000","tz":330},"subject":"Create stable/zed for tripleo repos","message":"Create stable/zed for tripleo repos\n\nAfter moving to independent release model, we did releases\nfor Xena and Yoga without stable branches. For Zed we\u0027re\ncreating stable/zed branch along with releases for all\nrelevant TripleO repos.\n\nChange-Id: Id6fa1b598ee430a6b496cf4f315e5fb3e7a2df4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d53c5e3e96588355fe5f8635ea31e4157b82bee2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d53c5e3e96588355fe5f8635ea31e4157b82bee2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[]}
