)]}'
{"id":"openstack%2Freleases~863182","triplet_id":"openstack%2Freleases~master~I0df8062bd800164d95e4ac9da0b0134d41accc48","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-11-07 10:55:38.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-11-07 10:55:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0df8062bd800164d95e4ac9da0b0134d41accc48","subject":"[manila] New release for Manila/Zed","status":"MERGED","created":"2022-11-01 21:11:30.000000000","updated":"2022-11-07 10:55:38.000000000","submitted":"2022-11-07 10:55:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863182","meta_rev_id":"34a6b39633de2fdcb48f3a2e82566797d10f26ad","_number":863182,"virtual_id_number":863182,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-07 10:55:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2022-11-03 13:20:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2022-11-03 13:58:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2022-11-07 10:43:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2022-11-07 10:43:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-11-01 21:49:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-01 21:12:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-01 21:20:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-11-03 13:58:47.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2022-11-07 10:43:04.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"82732f08d5612ae88c347083646baa50c7dc6a00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-11-01 21:11:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"def526872556853b4cd09228fffa5b94c4d7eaea","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-01 21:12:21.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"659eb3f39637beb605ce6e941a89465d0414318a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-01 21:23:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e31b7d6f82534f4c8295e95249ee28c4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d44f6248db0746fab385b0bde46a6e64 : SUCCESS in 6m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8c68a6009a124b8793fda15821dbca2c : SUCCESS in 9m 33s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e19f960efa97458495ff2f02d4fa36ac : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":1},{"id":"eddb16e322f88d6e03af1aa4995c068b88d7e066","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-11-01 21:46:48.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7c135af3fa2d91305b4c38af904693aedf9a6f82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-11-01 21:48:38.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b1f589d8c2a3cc9fe49e5bed82c48cb88260d7d9","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-01 21:49:26.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"8dd6a67f3fb6a8684b88b3c3abf9d8dc3c66ecd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-01 22:06:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd55fa74770a410ea898974cfb1879ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/587fa5d68fb44cde92d3436bf25049d6 : SUCCESS in 5m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/debd0bd160744761ae0d226920b7619e : SUCCESS in 5m 39s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4a8728290eed449db4c4bf6f531b61cf : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":2},{"id":"aa204c7186e835a5923a9950e152a190e4b8d3c9","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-11-03 13:20:35.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cb4200f8912e4502adf9e312d3edf2d763fc90e3","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-11-03 13:58:47.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7c3da939c498e316e3017437deceee60edb301bb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-07 10:43:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2b3a5c4e765253e1a7d6c64155e5fd1b2d342da8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 10:43:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2c3bcd861103d441e43bb458eec632e529b52585","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 10:55:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40c13806f1b44afa97d02f5fee38256e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2246ad7941f14e989b0f835ac9d4b960 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d58e32d084ad4ecabf16c67aaff5bccb : SUCCESS in 6m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/77f146901155417e9dd38b27609f19cf : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":2},{"id":"34a6b39633de2fdcb48f3a2e82566797d10f26ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 10:55:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"63683c179e70557749212f11b10ebcc25e1da3e0","revisions":{"8c1149e29ff3273011b07082f171711156a01513":{"kind":"REWORK","_number":1,"created":"2022-11-01 21:11:30.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/82/863182/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/82/863182/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/82/863182/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/82/863182/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/82/863182/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/82/863182/1"}}},"commit":{"parents":[{"commit":"d6031738ef67a6d7741f353413e1dd63b28e6f4c","subject":"Merge \"[zaqar] Transition Wallaby to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6031738ef67a6d7741f353413e1dd63b28e6f4c"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-11-01 21:08:57.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-11-01 21:08:57.000000000","tz":-180},"subject":"[manila] New release for Manila/Zed","message":"[manila] New release for Manila/Zed\n\nA couple of bugfixes were added to the stable Zed branch. Fixes\ninclude:\n\ngit log --oneline --no-merges 15.0.0..2e16e27\n5f9c037a3 Set configured_ip_versions fact in cephfs driver\ncfdb3c432 Macrosan driver: some improvements about variables and syntax\ndf2cc333e Fix neutron plugin get network_type\n440881392 Allow setting endpoint_type for glance\n6a3f9f06c Rename \u0027nfs_cluster_id\u0027 to \u0027cluster_id\u0027 in the NFSProtocolHelper\n4fd8e223e [CI][stable/zed] pin test plugin in grenade skip-level\n078bc4300 [RBAC] Return 404 if share is inaccessible\ndea2503c8 (gerrit_ssh/stable/zed) Update TOX_CONSTRAINTS_FILE for stable/zed\n18fa883b2 Update .gitreview for stable/zed\n\nChange-Id: I0df8062bd800164d95e4ac9da0b0134d41accc48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8c1149e29ff3273011b07082f171711156a01513"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8c1149e29ff3273011b07082f171711156a01513"}]},"branch":"refs/heads/master"},"63683c179e70557749212f11b10ebcc25e1da3e0":{"kind":"REWORK","_number":2,"created":"2022-11-01 21:48:38.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/82/863182/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/82/863182/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/82/863182/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/82/863182/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/82/863182/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/82/863182/2"}}},"commit":{"parents":[{"commit":"d6031738ef67a6d7741f353413e1dd63b28e6f4c","subject":"Merge \"[zaqar] Transition Wallaby to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d6031738ef67a6d7741f353413e1dd63b28e6f4c"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-11-01 21:08:57.000000000","tz":-180},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-11-01 21:48:37.000000000","tz":0},"subject":"[manila] New release for Manila/Zed","message":"[manila] New release for Manila/Zed\n\nSeveral bugfixes were added to the stable Zed branch. \nTwo new configuration options were added as well [1][2], allowing\nfor new behavior by preserving backwards compatibility.\n\nFixes include:\n\ngit log --oneline --no-merges 15.0.0..2e16e27\n5f9c037a3 Set configured_ip_versions fact in cephfs driver\ncfdb3c432 Macrosan driver: some improvements about variables and syntax   [1]\ndf2cc333e Fix neutron plugin get network_type\n440881392 Allow setting endpoint_type for glance                          [2]\n6a3f9f06c Rename \u0027nfs_cluster_id\u0027 to \u0027cluster_id\u0027 in the NFSProtocolHelper\n4fd8e223e [CI][stable/zed] pin test plugin in grenade skip-level\n078bc4300 [RBAC] Return 404 if share is inaccessible\ndea2503c8 (gerrit_ssh/stable/zed) Update TOX_CONSTRAINTS_FILE for stable/zed\n18fa883b2 Update .gitreview for stable/zed\n\nChange-Id: I0df8062bd800164d95e4ac9da0b0134d41accc48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/63683c179e70557749212f11b10ebcc25e1da3e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/63683c179e70557749212f11b10ebcc25e1da3e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[]}
