)]}'
{"id":"openstack%2Freleases~863695","triplet_id":"openstack%2Freleases~master~Ie3e9ddb610124f7eaea76e71a1aae3fe7b13cf4e","project":"openstack/releases","branch":"master","topic":"pycadf","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2022-11-04 20:06:35.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie3e9ddb610124f7eaea76e71a1aae3fe7b13cf4e","subject":"Move stable/wallaby in pycadf to EM","status":"ABANDONED","created":"2022-11-04 19:09:33.000000000","updated":"2022-11-04 20:06:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b7395c9698c277d035b552168d1081015f2af0af","_number":863695,"virtual_id_number":863695,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-11-04 19:21:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":1,"date":"2022-11-04 19:09:53.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-11-04 19:10:58.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-04 19:09:53.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2022-11-04 19:10:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4e0351c0750982a0f5dfafd97b5ac2d3f2d6ce9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-11-04 19:09:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"778433a7a77d7f33c96f37a5ba4ae42940a808c8","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-11-04 19:09:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a6ed7304676db7013084e56dbd7f9dcfbd37edcd","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 19:10:58.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"659349117e164baccf791e389849e0b0e1047af3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 19:21:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/43b3f72ab27c48b097b4d3947a9a191c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4b121413d494bc4aabe59778c4cfc95 : SUCCESS in 6m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/473ba30d1bb94a499ae0846725de2b89 : FAILURE in 4m 32s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f57d50eb318a4d0bb1ff066109bf047e : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b7395c9698c277d035b552168d1081015f2af0af","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-11-04 20:06:35.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"610054c54bf94a27db912a1fd42179ae623cea78","revisions":{"610054c54bf94a27db912a1fd42179ae623cea78":{"kind":"REWORK","_number":1,"created":"2022-11-04 19:09:33.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/95/863695/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/863695/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/863695/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/863695/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/863695/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/863695/1"}}},"commit":{"parents":[{"commit":"5ba7c017e636c33532accc22b099fb33fcf2630a","subject":"Add stable/wallaby branch for pycadf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5ba7c017e636c33532accc22b099fb33fcf2630a"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-11-04 19:08:53.000000000","tz":-240},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-11-04 19:08:53.000000000","tz":-240},"subject":"Move stable/wallaby in pycadf to EM","message":"Move stable/wallaby in pycadf to EM\n\nChange-Id: Ie3e9ddb610124f7eaea76e71a1aae3fe7b13cf4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/610054c54bf94a27db912a1fd42179ae623cea78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/610054c54bf94a27db912a1fd42179ae623cea78"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
