)]}'
{"id":"openstack%2Freleases~863698","triplet_id":"openstack%2Freleases~master~Ib3640147c3fceee6b915cf7d3794afcf4b02bf49","project":"openstack/releases","branch":"master","topic":"pycadf_xena","attention_set":{},"removed_from_attention_set":{"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2022-11-04 20:03:55.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib3640147c3fceee6b915cf7d3794afcf4b02bf49","subject":"Release pycadf for xena","status":"ABANDONED","created":"2022-11-04 19:45:11.000000000","updated":"2022-11-04 20:03:55.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a6885102254cb6f705d778c166b733b7975f96ca","_number":863698,"virtual_id_number":863698,"owner":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-11-04 19:54:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-11-04 19:48:38.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-04 19:46:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"304535c493d8fd5038881a77cd20ec1abbc48f15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-11-04 19:45:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9702030af038f51b9690eb5a47c72b61f0feafc3","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 19:46:00.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"34065f5bcb267b028896bbdbbae4af1bfce30c34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-11-04 19:46:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6760a0a89245c6bbee7c3032685772272b02badb","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 19:47:40.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"3df1872d8f315cf3d904f0c570c13b2cd796cae7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-11-04 19:47:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c043948215af6acf0bdc2d2f723a250c939b91c4","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 19:48:38.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"e0fa6926b58f3edd2aa0bf0a51514dde98c0e65c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 19:54:39.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/22e402a7d9b84a6e9c9108990c167f7e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/516153e473944e7f9af21b93c472addf : SUCCESS in 5m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0dc8297b33234118a69d9be48ade51c0 : FAILURE in 3m 23s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d179a09679a041088338b9ae9fa160fa : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":3},{"id":"a6885102254cb6f705d778c166b733b7975f96ca","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-11-04 20:03:55.000000000","message":"Abandoned\n\nIndependent Release","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6ec7ed9c3237a68e251ae7f79af9f71a8605215b","revisions":{"b430f2818d7f3352572d1307d7e1d3a57ef77bfd":{"kind":"REWORK","_number":1,"created":"2022-11-04 19:45:11.000000000","uploader":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"ref":"refs/changes/98/863698/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/98/863698/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/863698/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/863698/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/863698/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/98/863698/1"}}},"commit":{"parents":[{"commit":"6e8b11ba0db515dc79f880186b129460c752c2d8","subject":"[nova] Transition Wallaby to EM","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6e8b11ba0db515dc79f880186b129460c752c2d8"}]}],"author":{"name":"Dave Wilde (d34dh0r53)","email":"dwilde@redhat.com","date":"2022-11-04 19:44:21.000000000","tz":-300},"committer":{"name":"Dave Wilde (d34dh0r53)","email":"dwilde@redhat.com","date":"2022-11-04 19:45:10.000000000","tz":-300},"subject":"Release pycadf for xena","message":"Release pycadf for xena\n\npycadf seems to have been left behind since Victoria.  We need this\nrelease so we can branch stable/xena.\n\nChange-Id: Ib3640147c3fceee6b915cf7d3794afcf4b02bf49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b430f2818d7f3352572d1307d7e1d3a57ef77bfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b430f2818d7f3352572d1307d7e1d3a57ef77bfd"}]},"branch":"refs/heads/master"},"2274efd5181f114ffd9e1b615add76849699d81f":{"kind":"REWORK","_number":2,"created":"2022-11-04 19:46:45.000000000","uploader":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"ref":"refs/changes/98/863698/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/98/863698/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/863698/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/863698/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/863698/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/98/863698/2"}}},"commit":{"parents":[{"commit":"6e8b11ba0db515dc79f880186b129460c752c2d8","subject":"[nova] Transition Wallaby to EM","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6e8b11ba0db515dc79f880186b129460c752c2d8"}]}],"author":{"name":"Dave Wilde (d34dh0r53)","email":"dwilde@redhat.com","date":"2022-11-04 19:44:21.000000000","tz":-300},"committer":{"name":"Dave Wilde (d34dh0r53)","email":"dwilde@redhat.com","date":"2022-11-04 19:46:36.000000000","tz":-300},"subject":"Release pycadf for xena","message":"Release pycadf for xena\n\npycadf seems to have been left behind since Victoria.  We need this\nrelease so we can branch stable/xena.\n\nChange-Id: Ib3640147c3fceee6b915cf7d3794afcf4b02bf49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2274efd5181f114ffd9e1b615add76849699d81f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2274efd5181f114ffd9e1b615add76849699d81f"}]},"branch":"refs/heads/master"},"6ec7ed9c3237a68e251ae7f79af9f71a8605215b":{"kind":"REWORK","_number":3,"created":"2022-11-04 19:47:53.000000000","uploader":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"ref":"refs/changes/98/863698/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/98/863698/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/863698/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/863698/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/863698/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/98/863698/3"}}},"commit":{"parents":[{"commit":"6e8b11ba0db515dc79f880186b129460c752c2d8","subject":"[nova] Transition Wallaby to EM","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6e8b11ba0db515dc79f880186b129460c752c2d8"}]}],"author":{"name":"Dave Wilde (d34dh0r53)","email":"dwilde@redhat.com","date":"2022-11-04 19:44:21.000000000","tz":-300},"committer":{"name":"Dave Wilde (d34dh0r53)","email":"dwilde@redhat.com","date":"2022-11-04 19:47:45.000000000","tz":-300},"subject":"Release pycadf for xena","message":"Release pycadf for xena\n\npycadf seems to have been left behind since Victoria.  We need this\nrelease so we can branch stable/xena.\n\nChange-Id: Ib3640147c3fceee6b915cf7d3794afcf4b02bf49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6ec7ed9c3237a68e251ae7f79af9f71a8605215b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6ec7ed9c3237a68e251ae7f79af9f71a8605215b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
