)]}'
{"id":"openstack%2Freleases~863700","triplet_id":"openstack%2Freleases~master~I553f0746eb7f16cfb1695ff15b4aec86d8e96662","project":"openstack/releases","branch":"master","topic":"pycadf_xena","attention_set":{},"removed_from_attention_set":{"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2022-11-04 20:04:20.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I553f0746eb7f16cfb1695ff15b4aec86d8e96662","subject":"Add stable/xena branch for pycadf","status":"ABANDONED","created":"2022-11-04 19:48:52.000000000","updated":"2022-11-04 20:04:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"95f73cd7828bf9b83fedb8dc7817d28d9fa6d1bc","_number":863700,"virtual_id_number":863700,"owner":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-11-04 20:03:20.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-04 19:49:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f63be738e5e3a4a3f31f053def1ffea6fe8d9316","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-11-04 19:48:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c4eade3acc577ba1ff4e6cd0c1cbc86582ed444","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 19:49:43.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"76aa97de1a0302d228122bb83a1934905f706a26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 19:57:14.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/193a009e737243a098c98d3be64b40fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb9ca368369b49c69b7249e292edd51a : SUCCESS in 6m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0e4953deb085455a9ac3fe341a3fd748 : FAILURE in 4m 16s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ecfec81a5a9b4558bf8aced4b0d6b485 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":1},{"id":"b52ff856a8d883a5d7bad0871341e68bdb273808","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-11-04 20:02:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8a71691b7c7251f76f984cac7d18ad4822393e57","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 20:03:20.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"95f73cd7828bf9b83fedb8dc7817d28d9fa6d1bc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-11-04 20:04:20.000000000","message":"Abandoned\n\nIndependent Release","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2ec8300698e6fda8992a7a5a3b6e0934aade1506","revisions":{"b071d78aa93943ee152d2717684e1c842dc74779":{"kind":"REWORK","_number":1,"created":"2022-11-04 19:48:52.000000000","uploader":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"ref":"refs/changes/00/863700/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/00/863700/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/863700/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/863700/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/863700/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/00/863700/1"}}},"commit":{"parents":[{"commit":"6ec7ed9c3237a68e251ae7f79af9f71a8605215b","subject":"Release pycadf for xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6ec7ed9c3237a68e251ae7f79af9f71a8605215b"}]}],"author":{"name":"Dave Wilde (d34dh0r53)","email":"dwilde@redhat.com","date":"2022-11-04 19:48:08.000000000","tz":-300},"committer":{"name":"Dave Wilde (d34dh0r53)","email":"dwilde@redhat.com","date":"2022-11-04 19:48:08.000000000","tz":-300},"subject":"Add stable/xena branch for pycadf","message":"Add stable/xena branch for pycadf\n\nThis patch adds the missing stable/wallaby branch at 3.3.0.\n\nChange-Id: I553f0746eb7f16cfb1695ff15b4aec86d8e96662\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b071d78aa93943ee152d2717684e1c842dc74779"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b071d78aa93943ee152d2717684e1c842dc74779"}]},"branch":"refs/heads/master"},"2ec8300698e6fda8992a7a5a3b6e0934aade1506":{"kind":"REWORK","_number":2,"created":"2022-11-04 20:02:26.000000000","uploader":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"ref":"refs/changes/00/863700/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/00/863700/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/863700/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/863700/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/863700/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/00/863700/2"}}},"commit":{"parents":[{"commit":"6ec7ed9c3237a68e251ae7f79af9f71a8605215b","subject":"Release pycadf for xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6ec7ed9c3237a68e251ae7f79af9f71a8605215b"}]}],"author":{"name":"Dave Wilde (d34dh0r53)","email":"dwilde@redhat.com","date":"2022-11-04 19:48:08.000000000","tz":-300},"committer":{"name":"Dave Wilde (d34dh0r53)","email":"dwilde@redhat.com","date":"2022-11-04 20:02:06.000000000","tz":-300},"subject":"Add stable/xena branch for pycadf","message":"Add stable/xena branch for pycadf\n\nThis patch adds the missing stable/xena branch at 3.3.0.\n\nChange-Id: I553f0746eb7f16cfb1695ff15b4aec86d8e96662\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2ec8300698e6fda8992a7a5a3b6e0934aade1506"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2ec8300698e6fda8992a7a5a3b6e0934aade1506"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
