)]}'
{"id":"openstack%2Freleases~863703","triplet_id":"openstack%2Freleases~master~I1096fc509b6502f46b8728db46527a0369b18e8a","project":"openstack/releases","branch":"master","topic":"pycadf_yoga","attention_set":{},"removed_from_attention_set":{"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2022-11-04 20:04:12.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I1096fc509b6502f46b8728db46527a0369b18e8a","subject":"Release pycadf for yoga","status":"ABANDONED","created":"2022-11-04 19:51:11.000000000","updated":"2022-11-04 20:04:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d4eb7358850be35e8c1a2e348dbffe5200723a53","_number":863703,"virtual_id_number":863703,"owner":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-11-04 20:02:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-11-04 19:52:00.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-04 19:52:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"38e5137c79b6b63f45c8bb4d4f05f460f8ad8718","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-11-04 19:51:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"992c4448aa9c08812b394660f7f8b184a202d2ff","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 19:52:00.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"fec46b8c7168bfa8e051aeb0ece0fef6fd983b3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 20:02:26.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb8a0bd3b0194eaf8f12c54a4402cc75\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e17606947ea43b68961a71ae993fd19 : SUCCESS in 6m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9ffd217882e74c5286b628681c6d8772 : FAILURE in 3m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6b12805a1a2f44eaa0842263776e0d52 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":1},{"id":"d4eb7358850be35e8c1a2e348dbffe5200723a53","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-11-04 20:04:12.000000000","message":"Abandoned\n\nIndependent Release","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"98828dc93c79001e19ecef1a4b95f2fd3742c6d3","revisions":{"98828dc93c79001e19ecef1a4b95f2fd3742c6d3":{"kind":"REWORK","_number":1,"created":"2022-11-04 19:51:11.000000000","uploader":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"ref":"refs/changes/03/863703/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/03/863703/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/863703/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/863703/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/863703/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/03/863703/1"}}},"commit":{"parents":[{"commit":"6e8b11ba0db515dc79f880186b129460c752c2d8","subject":"[nova] Transition Wallaby to EM","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6e8b11ba0db515dc79f880186b129460c752c2d8"}]}],"author":{"name":"Dave Wilde (d34dh0r53)","email":"dwilde@redhat.com","date":"2022-11-04 19:50:30.000000000","tz":-300},"committer":{"name":"Dave Wilde (d34dh0r53)","email":"dwilde@redhat.com","date":"2022-11-04 19:50:30.000000000","tz":-300},"subject":"Release pycadf for yoga","message":"Release pycadf for yoga\n\npycadf seems to have been left behind since Victoria.  We need this\nrelease so we can branch stable/yoga.\n\nChange-Id: I1096fc509b6502f46b8728db46527a0369b18e8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/98828dc93c79001e19ecef1a4b95f2fd3742c6d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/98828dc93c79001e19ecef1a4b95f2fd3742c6d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
