)]}'
{"id":"openstack%2Freleases~864066","triplet_id":"openstack%2Freleases~master~Iaf476c80eed5c71316764890b40a631e1df12758","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-11-10 10:03:11.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-11-10 10:03:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaf476c80eed5c71316764890b40a631e1df12758","subject":"Release manila-tempest-plugin","status":"MERGED","created":"2022-11-08 21:37:23.000000000","updated":"2022-11-10 10:03:11.000000000","submitted":"2022-11-10 10:03:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864066","meta_rev_id":"aeb6679f2ce9d69a9d8303607b522d07785eedaa","_number":864066,"virtual_id_number":864066,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-10 10:03:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2022-11-10 08:51:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2022-11-10 09:57:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2022-11-10 09:57:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-11-09 20:14:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-08 21:38:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-08 22:20:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-11-10 08:51:11.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2022-11-10 09:57:01.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"604bcc0d0236817f2d1210a2bafa4f7b29e3bf79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-11-08 21:37:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aba470ab1628e010a9ca955407519fea73b68384","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 21:38:16.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"eee079022dfc1c7ee0f424f6626e9e0cc54fb91f","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-11-08 21:38:17.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"32afde60367efa865542d8dda968870e8a68b30f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 21:52:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80a21d11436d4d3c9a3b1b2ce7c882cd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/078281469ae44da3bd0ccc49e4ea1a0c : SUCCESS in 4m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cc47788d866c4eac8b2a308de429fdac : SUCCESS in 4m 15s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/75222f153cf4457798b2061fc908a35b : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"bb2cf99c154b6fed838b2d12c0705bfef7a109bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-11-09 20:14:09.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"154d8f1c36219051d501a47bf99a6c7133effb82","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-11-09 20:14:43.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b85a2b911d7c17c3d97c69f70fdc94574accbdd4","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 20:14:58.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"01184661cc00afc61e9d51bb6e978f5b75bc0ed9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 20:28:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca53c9b2e72c4eef9f05e3b99009f889\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1514af141264115a6bb39862434dcbd : SUCCESS in 6m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2bf346081d9841559301fae8be35c624 : SUCCESS in 4m 02s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1b919846e2934ab5a2c93096b0719fcb : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":2},{"id":"3afd5ba214f7fd233e459d5997d1df218e41c42b","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-11-10 08:51:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e0279c30072a32f66f776383b434338fd3c2a9b2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-10 09:57:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"90bfd0dbd6ee61dd73c19ed6506aeea63c4329ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 09:57:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2d85ff9cedb592917702dce4dd6cfa5201f190ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 10:03:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b11fbf753e8745a684552f2689238add\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3abf51670b314484abaa76555a48c38f : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/450656b5fce646f591590a65a7a90190 : SUCCESS in 5m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/95776de23db5421f8392a4acf94ca4e9 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":2},{"id":"aeb6679f2ce9d69a9d8303607b522d07785eedaa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 10:03:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e79fa6217de45517a797c8c1d03a1dc23dc1deff","revisions":{"993f84607bb8b92cf7d07779fae3bc76f0eefb19":{"kind":"REWORK","_number":1,"created":"2022-11-08 21:37:23.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/66/864066/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/66/864066/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/864066/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/864066/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/864066/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/66/864066/1"}}},"commit":{"parents":[{"commit":"c01e19ca96d17ec1c50ce15512e4584101a226a1","subject":"Merge \"[octavia] Transition Wallaby to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c01e19ca96d17ec1c50ce15512e4584101a226a1"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-11-08 21:36:26.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-11-08 21:36:26.000000000","tz":-180},"subject":"Release manila-tempest-plugin","message":"Release manila-tempest-plugin\n\nCouple of RBAC tests were included in the manila-tempest-plugin\nrepository, alongside some good fixes:\n\n$ git log --oneline --no-merges 1.10.0..256a6b0\na325504 Remove \"alt_project_share_v2_client\" client initialization\n3d2b7bb More Snapshot metadata field verification \u003e\u003d 2.73\n75dbca6 Add a metadata field shown on API \u003e\u003d 2.73\n7b2566a Add S-RBAC tests for manila\n5e559df Move skip condition under skip_checks class method\n8125ca4 Replace period to underscore\n\nChange-Id: Iaf476c80eed5c71316764890b40a631e1df12758\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/993f84607bb8b92cf7d07779fae3bc76f0eefb19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/993f84607bb8b92cf7d07779fae3bc76f0eefb19"}]},"branch":"refs/heads/master"},"e79fa6217de45517a797c8c1d03a1dc23dc1deff":{"kind":"REWORK","_number":2,"created":"2022-11-09 20:14:09.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/66/864066/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/66/864066/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/864066/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/864066/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/864066/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/66/864066/2"}}},"commit":{"parents":[{"commit":"c01e19ca96d17ec1c50ce15512e4584101a226a1","subject":"Merge \"[octavia] Transition Wallaby to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c01e19ca96d17ec1c50ce15512e4584101a226a1"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-11-08 21:36:26.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2022-11-09 20:14:02.000000000","tz":0},"subject":"Release manila-tempest-plugin","message":"Release manila-tempest-plugin\n\nCouple of RBAC tests were included in the manila-tempest-plugin\nrepository, alongside some good fixes:\n\n$ git log --oneline --no-merges 1.10.0..121f536\n121f536 (HEAD, origin/master, origin/HEAD) [RBAC] Fix test tags\na325504 Remove \"alt_project_share_v2_client\" client initialization\n3d2b7bb More Snapshot metadata field verification \u003e\u003d 2.73\n75dbca6 Add a metadata field shown on API \u003e\u003d 2.73\n7b2566a Add S-RBAC tests for manila\n5e559df Move skip condition under skip_checks class method\n8125ca4 Replace period to underscore\n\nChange-Id: Iaf476c80eed5c71316764890b40a631e1df12758\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e79fa6217de45517a797c8c1d03a1dc23dc1deff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e79fa6217de45517a797c8c1d03a1dc23dc1deff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[]}
