)]}'
{"id":"openstack%2Freleases~864095","triplet_id":"openstack%2Freleases~master~Icc9709c465e08b78650dc2a49a185c7228ecdc1e","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-11-17 13:30:49.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-11-17 13:40:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icc9709c465e08b78650dc2a49a185c7228ecdc1e","subject":"Moving back oslo deliverables to the CWI model","status":"MERGED","created":"2022-11-09 08:51:47.000000000","updated":"2022-11-17 13:40:01.000000000","submitted":"2022-11-17 13:40:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864095","meta_rev_id":"83b1fc84c61a9701ba60ee46f2349b21d87ae273","_number":864095,"virtual_id_number":864095,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-17 13:40:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2022-11-17 09:57:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2022-11-17 13:30:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2022-11-17 13:30:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-11-10 09:45:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-09 08:53:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-10 08:53:10.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"CC"},{"updated":"2022-11-10 14:05:49.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2022-11-11 08:42:26.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2022-11-17 09:57:23.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2022-11-17 13:30:49.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"a08fc5f11bb2df426b4425ea8ceedcee1337bdc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-09 08:51:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b9f31b9d0dcdebeb871d07dbce7f83e4e1a6fff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-09 08:52:23.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c89c4702e7bd2df608677840c76d01c449c0237e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 08:53:14.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"9b4db979df82989a8cfa7f9365981ee400971b23","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-09 08:53:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7452291adb338b5cee5a8481c276896588b8a552","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 09:01:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2902924179e4140b6eb81a24da553a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb9ed25dd947459783ef279e38320bbc : FAILURE in 3m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/502aa87ff4f645e7a8b991a99b68af6e : FAILURE in 3m 39s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3e487e22bde64be7a75d988c3e0bc7ea : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":2},{"id":"8285cf86b66f7f72163aa209a6b587a9f5c5d717","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-11-10 08:53:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"241fd8571e401869488ce23cfca1fff61dadd9b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-10 09:44:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d47963de24accaa1bfd0478a313a99ea018ba7e5","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 09:45:37.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"dac0b32db7f4d2f491f87b5e4e6e733c9e50b6f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 09:55:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b39432538f7f44f28bf56e52a9816080\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2f48f45c9574d0394ccce50186f48dd : SUCCESS in 6m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9b2049fc7a6843b6b8b396ddd7133c21 : SUCCESS in 4m 39s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/adec9cb5cbcd4af98bb4a6589c6dad7d : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":3},{"id":"9348f7a76b95f5a337342aceb88d4d658316a4f0","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-11-10 14:05:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f35505e78d2fec00acacc04fab25f6141acd5be7","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-11-17 09:57:23.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"aa1063e983518ea807fb6f65f2e39c5b3f0b4dd9","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-11-17 13:30:49.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5e4a54dc3b3c4c70d5147b829530d97b789b917d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 13:31:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c90e750da258d59a42644a946c012a6c31d90e46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 13:40:01.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/865b0dec049a410ebc3380b2c054f6eb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6299548ef0b45b1a39c8d7378843294 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed7ded2181c04d7eb31464ad917577df : SUCCESS in 6m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c5a68e7a94324f6494f52fdacd98e4a1 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":3},{"id":"83b1fc84c61a9701ba60ee46f2349b21d87ae273","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 13:40:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"782d8cbaf4e06f2b3663fcd8945d4e4ef7892507","revisions":{"6bf579e30fd612f0640fbb1079483c9932a696fc":{"kind":"REWORK","_number":1,"created":"2022-11-09 08:51:47.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/95/864095/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/864095/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/864095/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/864095/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/864095/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/864095/1"}}},"commit":{"parents":[{"commit":"1e9695c64134b78215f67eade8caf367b7cc3199","subject":"Merge \"[oslo] Transition Wallaby to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1e9695c64134b78215f67eade8caf367b7cc3199"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-11-09 08:43:58.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-11-09 08:43:58.000000000","tz":60},"subject":"Moving back oslo deliverables to coordinated releases","message":"Moving back oslo deliverables to coordinated releases\n\nFew cycles ago we moved a couple of oslo deliverables to the independent\nrelease model [1][2], however it led us to issues to backport patches\n[3].\n\nIndeed we made independent but want a backport after all.\nBackports are no longer an option for those deliverables.\n\nDuring the previous release PTG we decided [4] to move back the oslo\ndeliverables to the cycle-with-intermediary model.\n\n[1] https://lists.openstack.org/pipermail/openstack-discuss/2020-November/018527.html\n[2] https://opendev.org/openstack/releases/commit/5ecb80c82ed3ab0144c8e5860ee62df458dfc2b5\n[3] https://lists.openstack.org/pipermail/openstack-discuss/2022-September/030612.html\n[4] https://etherpad.opendev.org/p/oct2022-ptg-rel-mgt\n\nChange-Id: Icc9709c465e08b78650dc2a49a185c7228ecdc1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6bf579e30fd612f0640fbb1079483c9932a696fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6bf579e30fd612f0640fbb1079483c9932a696fc"}]},"branch":"refs/heads/master"},"f1dedeff576e8b377c869e34d59398b926a8b8aa":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-11-09 08:52:23.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/95/864095/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/864095/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/864095/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/864095/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/864095/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/864095/2"}}},"commit":{"parents":[{"commit":"1e9695c64134b78215f67eade8caf367b7cc3199","subject":"Merge \"[oslo] Transition Wallaby to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1e9695c64134b78215f67eade8caf367b7cc3199"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-11-09 08:43:58.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-11-09 08:52:01.000000000","tz":60},"subject":"Moving back oslo deliverables to the CWI model","message":"Moving back oslo deliverables to the CWI model\n\nFew cycles ago we moved a couple of oslo deliverables to the independent\nrelease model [1][2], however it led us to issues to backport patches\n[3].\n\nIndeed we made independent but want a backport after all.\nBackports are no longer an option for those deliverables.\n\nDuring the previous release PTG we decided [4] to move back the oslo\ndeliverables to the cycle-with-intermediary model.\n\n[1] https://lists.openstack.org/pipermail/openstack-discuss/2020-November/018527.html\n[2] https://opendev.org/openstack/releases/commit/5ecb80c82ed3ab0144c8e5860ee62df458dfc2b5\n[3] https://lists.openstack.org/pipermail/openstack-discuss/2022-September/030612.html\n[4] https://etherpad.opendev.org/p/oct2022-ptg-rel-mgt\n\nChange-Id: Icc9709c465e08b78650dc2a49a185c7228ecdc1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f1dedeff576e8b377c869e34d59398b926a8b8aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f1dedeff576e8b377c869e34d59398b926a8b8aa"}]},"branch":"refs/heads/master"},"782d8cbaf4e06f2b3663fcd8945d4e4ef7892507":{"kind":"REWORK","_number":3,"created":"2022-11-10 09:44:45.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/95/864095/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/864095/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/864095/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/864095/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/864095/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/864095/3"}}},"commit":{"parents":[{"commit":"1e9695c64134b78215f67eade8caf367b7cc3199","subject":"Merge \"[oslo] Transition Wallaby to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1e9695c64134b78215f67eade8caf367b7cc3199"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-11-09 08:43:58.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-11-10 09:44:34.000000000","tz":60},"subject":"Moving back oslo deliverables to the CWI model","message":"Moving back oslo deliverables to the CWI model\n\nFew cycles ago we moved a couple of oslo deliverables to the independent\nrelease model [1][2], however it led us to issues to backport patches\n[3].\n\nIndeed we made independent but want a backport after all.\nBackports are no longer an option for those deliverables.\n\nDuring the previous release PTG we decided [4] to move back the oslo\ndeliverables to the cycle-with-intermediary model.\n\n[1] https://lists.openstack.org/pipermail/openstack-discuss/2020-November/018527.html\n[2] https://opendev.org/openstack/releases/commit/5ecb80c82ed3ab0144c8e5860ee62df458dfc2b5\n[3] https://lists.openstack.org/pipermail/openstack-discuss/2022-September/030612.html\n[4] https://etherpad.opendev.org/p/oct2022-ptg-rel-mgt\n\nChange-Id: Icc9709c465e08b78650dc2a49a185c7228ecdc1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/782d8cbaf4e06f2b3663fcd8945d4e4ef7892507"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/782d8cbaf4e06f2b3663fcd8945d4e4ef7892507"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[]}
