)]}'
{"id":"openstack%2Freleases~864512","triplet_id":"openstack%2Freleases~master~I43fdb4cf5582097fd6d89c2d8f04844d31e275c9","project":"openstack/releases","branch":"master","topic":"zed-trailing-branch-cut","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-12-13 08:57:02.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-12-13 12:59:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I43fdb4cf5582097fd6d89c2d8f04844d31e275c9","subject":"[OpenStackAnsible] Release openstack-ansible-roles for Zed and cut branch","status":"MERGED","created":"2022-11-15 10:32:36.000000000","updated":"2022-12-13 12:59:58.000000000","submitted":"2022-12-13 12:59:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864512-zed-trailing-branch-cut","meta_rev_id":"a76c3931f4b3f59e29d5b49145bb4cfd5b31b3aa","_number":864512,"virtual_id_number":864512,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-13 12:59:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2022-12-13 12:53:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2022-12-13 08:57:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2022-12-13 12:53:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-12-13 03:45:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-15 10:33:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-15 10:50:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-15 12:20:09.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-12-09 13:12:52.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2022-12-13 12:53:37.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"7cdba87f07ab20c0e194ca09a8ae079fb82e6b7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-15 10:32:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30b04347c88ffb4bfc71b7c6b485537e05075720","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 10:50:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca268b53a3274c199232b71904bcb42e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/198886b6165d4ee8a59474025ee0d597 : SUCCESS in 5m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cd39326ec7b8499187dd9d45ea71f3e3 : SUCCESS in 5m 30s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b739da9e6b86431f8b56468ed4c44615 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"082a39f328b1efd80b1b2b4467c61862bff9124f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-17 12:22:06.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"b7e502b04f9dff6d8e4b1ec8e95585909950e382","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-12-09 13:12:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ac68fbf8778129dd2c81b287949a40aedf2a1c03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-12 20:29:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"75ef409c7047c3275df6a820770d2636c1421ef2","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 20:30:31.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"23483e7bfe71f18ba5e629e7f4beec529ebce0fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 20:36:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d34d29e0972c49cfa9b6f454046c109b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4abc67d3a854a5182ebc101ed8d60c8 : SUCCESS in 5m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c478cbe8bb1d4a1eb8f356addd5650ac : FAILURE in 4m 58s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c677b497ebd84124880036cc46ccd3b2 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":2},{"id":"45cb9630da4075273dc590cc334a52018de4a855","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-12 21:58:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"219c2dd5fbf75295368917273171f3e9e7332437","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 21:59:34.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"2449bb7f1396554e30f2df0f66b3b209646cb560","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 22:08:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/546303a2c8dd481f89b19d09dfd7616e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76cfefd8a31e49bcb7bb518537d8647d : SUCCESS in 5m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2f63b81c145042f192ab2172636db0f5 : FAILURE in 4m 16s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/62a0b3d34a5e4a88a1fadef4bd5fdfdb : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":3},{"id":"d61dbe170255ef1ffd335123bd988ccb3902010e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-13 03:43:46.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"7251919b199bcd6899f427e8bf6f346659814352","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 03:44:33.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"6deaed2c50639946f437123aef1480bd8cb91249","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-13 03:44:59.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"abf05a8eb43585ad3f59ea8a5c05c7d4830879e0","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 03:45:44.000000000","message":"Patch Set 5: PTL-Approved+1","accounts_in_message":[],"_revision_number":5},{"id":"7d11764fd86024da0e1916dc525d5791c15d98d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 03:56:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2db5284cc3347bca8b703e0e05286eb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b18c6e99b3ec43f4a444977cab9f1e14 : SUCCESS in 6m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/833d114943ac43ebbb3118636d6315c0 : SUCCESS in 5m 08s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/627df14314dd4dcea95210887a6141a2 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":5},{"id":"181066df41a0a9ca19b1d552ca9b8d565f24a953","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-12-13 08:57:02.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"bad0629bd5edce20dbb9611621e3965e1ba77452","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-12-13 12:53:37.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0f79a0d676bd355cadbff5dd075e6f38b5ceb912","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 12:53:58.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"aa5b7d7ef3a42b54c2aae4a3c3ac0184f20e9e18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 12:59:58.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e581863827714e25b585e66782fbeab5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fda99e70d244d8084627abd92a82bfe : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d74ff852b2244d22a5e768f17e42dc87 : SUCCESS in 5m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c1cdd8151b9943129155904599062e71 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":5},{"id":"a76c3931f4b3f59e29d5b49145bb4cfd5b31b3aa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 12:59:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"200a68635785bfc8b528f66864aaa00282e3ed2d","revisions":{"62be508b8670a501d39a2aaed971f6182716aa11":{"kind":"REWORK","_number":1,"created":"2022-11-15 10:32:36.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/12/864512/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/12/864512/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/864512/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/864512/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/864512/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/12/864512/1"}}},"commit":{"parents":[{"commit":"403444806a946cd506770a5bc10b45a77172d40e","subject":"Merge \"[ironic] Transition Wallaby to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/403444806a946cd506770a5bc10b45a77172d40e"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-11-15 10:31:56.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-11-15 10:31:56.000000000","tz":60},"subject":"[OpenStackAnsible] Release openstack-ansible-roles for Zed and cut branch","message":"[OpenStackAnsible] Release openstack-ansible-roles for Zed and cut branch\n\nThis release patch is to ensure, that trailing typed projects have been\nbranched for Zed. The deadline for these deliverables is December 14.\n\nPlease +1 if the team is ready for us to proceed with this or update/-1\nif something should be changed.\n\nChange-Id: I43fdb4cf5582097fd6d89c2d8f04844d31e275c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/62be508b8670a501d39a2aaed971f6182716aa11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/62be508b8670a501d39a2aaed971f6182716aa11"}]},"branch":"refs/heads/master"},"18c357f232db53851c6b5eaed32687573fef3716":{"kind":"REWORK","_number":2,"created":"2022-12-12 20:29:42.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/12/864512/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/12/864512/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/864512/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/864512/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/864512/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/12/864512/2"}}},"commit":{"parents":[{"commit":"403444806a946cd506770a5bc10b45a77172d40e","subject":"Merge \"[ironic] Transition Wallaby to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/403444806a946cd506770a5bc10b45a77172d40e"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-11-15 10:31:56.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-12-12 20:29:34.000000000","tz":60},"subject":"[OpenStackAnsible] Release openstack-ansible-roles for Zed and cut branch","message":"[OpenStackAnsible] Release openstack-ansible-roles for Zed and cut branch\n\nThis release patch is to ensure, that trailing typed projects have been\nbranched for Zed. The deadline for these deliverables is December 14.\n\nPlease +1 if the team is ready for us to proceed with this or update/-1\nif something should be changed.\n\nChange-Id: I43fdb4cf5582097fd6d89c2d8f04844d31e275c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/18c357f232db53851c6b5eaed32687573fef3716"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/18c357f232db53851c6b5eaed32687573fef3716"}]},"branch":"refs/heads/master"},"01f98ba8bedacf417d2d41351c8458d9351f1a5e":{"kind":"REWORK","_number":3,"created":"2022-12-12 21:58:51.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/12/864512/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/12/864512/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/864512/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/864512/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/864512/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/12/864512/3"}}},"commit":{"parents":[{"commit":"403444806a946cd506770a5bc10b45a77172d40e","subject":"Merge \"[ironic] Transition Wallaby to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/403444806a946cd506770a5bc10b45a77172d40e"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-11-15 10:31:56.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-12-12 21:58:42.000000000","tz":60},"subject":"[OpenStackAnsible] Release openstack-ansible-roles for Zed and cut branch","message":"[OpenStackAnsible] Release openstack-ansible-roles for Zed and cut branch\n\nThis release patch is to ensure, that trailing typed projects have been\nbranched for Zed. The deadline for these deliverables is December 14.\n\nPlease +1 if the team is ready for us to proceed with this or update/-1\nif something should be changed.\n\nChange-Id: I43fdb4cf5582097fd6d89c2d8f04844d31e275c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/01f98ba8bedacf417d2d41351c8458d9351f1a5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/01f98ba8bedacf417d2d41351c8458d9351f1a5e"}]},"branch":"refs/heads/master"},"9fdbf253af7bfa7f76f924593e893d207bbc694f":{"kind":"REWORK","_number":4,"created":"2022-12-13 03:43:46.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/12/864512/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/12/864512/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/864512/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/864512/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/864512/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/12/864512/4"}}},"commit":{"parents":[{"commit":"403444806a946cd506770a5bc10b45a77172d40e","subject":"Merge \"[ironic] Transition Wallaby to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/403444806a946cd506770a5bc10b45a77172d40e"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-11-15 10:31:56.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-12-13 03:43:45.000000000","tz":0},"subject":"[OpenStackAnsible] Release openstack-ansible-roles for Zed and cut branch","message":"[OpenStackAnsible] Release openstack-ansible-roles for Zed and cut branch\n\nThis release patch is to ensure, that trailing typed projects have been\nbranched for Zed. The deadline for these deliverables is December 14.\n\nPlease +1 if the team is ready for us to proceed with this or update/-1\nif something should be changed.\n\nChange-Id: I43fdb4cf5582097fd6d89c2d8f04844d31e275c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9fdbf253af7bfa7f76f924593e893d207bbc694f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9fdbf253af7bfa7f76f924593e893d207bbc694f"}]},"branch":"refs/heads/master"},"200a68635785bfc8b528f66864aaa00282e3ed2d":{"kind":"REWORK","_number":5,"created":"2022-12-13 03:44:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/12/864512/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/12/864512/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/864512/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/864512/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/864512/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/12/864512/5"}}},"commit":{"parents":[{"commit":"403444806a946cd506770a5bc10b45a77172d40e","subject":"Merge \"[ironic] Transition Wallaby to EM\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/403444806a946cd506770a5bc10b45a77172d40e"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-11-15 10:31:56.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-12-13 03:44:58.000000000","tz":0},"subject":"[OpenStackAnsible] Release openstack-ansible-roles for Zed and cut branch","message":"[OpenStackAnsible] Release openstack-ansible-roles for Zed and cut branch\n\nThis release patch is to ensure, that trailing typed projects have been\nbranched for Zed. The deadline for these deliverables is December 14.\n\nPlease +1 if the team is ready for us to proceed with this or update/-1\nif something should be changed.\n\nChange-Id: I43fdb4cf5582097fd6d89c2d8f04844d31e275c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/200a68635785bfc8b528f66864aaa00282e3ed2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/200a68635785bfc8b528f66864aaa00282e3ed2d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[]}
