)]}'
{"id":"openstack%2Freleases~865577","triplet_id":"openstack%2Freleases~master~Ied73a19d6c733f32de397d8c2ad0b4d5547cb379","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"35600":{"account":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"last_update":"2023-01-30 06:29:16.000000000","reason":"Change was abandoned"},"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2023-01-30 06:29:16.000000000","reason":"Change was abandoned"},"15895":{"account":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"last_update":"2023-01-30 06:29:16.000000000","reason":"Change was abandoned"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-01-30 06:29:16.000000000","reason":"Change was abandoned"},"8731":{"account":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"last_update":"2023-01-30 06:29:16.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ied73a19d6c733f32de397d8c2ad0b4d5547cb379","subject":"Remove mistral deliverables from 2023.1 Antelope","status":"ABANDONED","created":"2022-11-24 15:50:49.000000000","updated":"2023-01-30 06:29:16.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"634d471a3642ddcf75900d85085aed18903096ce","_number":865577,"virtual_id_number":865577,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-11-24 15:59:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-24 15:51:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-24 15:54:05.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2022-11-24 15:54:05.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-11-24 15:59:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-24 16:20:14.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2022-11-24 16:20:15.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2023-01-28 12:24:35.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"}],"messages":[{"id":"38632b89a3512f7a943a98d524b202a9b6f93668","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-11-24 15:50:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"249fa0ab7720bf72b454861cc26f2fe0fb4e2f63","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-11-24 15:54:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c0130bdd09b35a61eadcd4054d0780d3241ca948","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 15:59:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ffac84f0a2c4268aa1e72f096975f56\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66e11c9e7a9b4fb9b035fadbfab78da7 : SUCCESS in 6m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/62fba108e2744b07b0d0fb7fbe431868 : SUCCESS in 4m 35s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b6ccadd363e14e8e8f0c1c90d9e17bd3 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"634d471a3642ddcf75900d85085aed18903096ce","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-01-30 06:29:16.000000000","message":"Abandoned\n\nmaintainers step up and fixed the gates, this can be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cb8b2bcea7f7e401ca5389ff3afbab7457fe211d","revisions":{"cb8b2bcea7f7e401ca5389ff3afbab7457fe211d":{"kind":"REWORK","_number":1,"created":"2022-11-24 15:50:49.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/77/865577/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/77/865577/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/77/865577/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/77/865577/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/77/865577/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/77/865577/1"}}},"commit":{"parents":[{"commit":"94a8a8144c7a5b414153e28a86f164de6ed00626","subject":"Release OpenStackSDK 0.103.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/94a8a8144c7a5b414153e28a86f164de6ed00626"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-11-24 15:46:02.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-11-24 15:50:10.000000000","tz":60},"subject":"Remove mistral deliverables from 2023.1 Antelope","message":"Remove mistral deliverables from 2023.1 Antelope\n\nMistral deliverables are not actively maintained, thus in release\nmanagement team we tend to have probelms with them at release\ndeadlines. I propose to remove these deliverables from official\n2023.1 Antelope release.\n\nChange-Id: Ied73a19d6c733f32de397d8c2ad0b4d5547cb379\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cb8b2bcea7f7e401ca5389ff3afbab7457fe211d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cb8b2bcea7f7e401ca5389ff3afbab7457fe211d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
