)]}'
{"id":"openstack%2Freleases~867174","triplet_id":"openstack%2Freleases~master~I871156f0e8e5ebb36bb695ffb9b10c491b355337","project":"openstack/releases","branch":"master","topic":"release-masakari-bugfix","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-12-12 13:30:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I871156f0e8e5ebb36bb695ffb9b10c491b355337","subject":"[masakari] Release bugfix","status":"MERGED","created":"2022-12-10 18:04:10.000000000","updated":"2022-12-12 13:30:55.000000000","submitted":"2022-12-12 13:30:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867174-release-masakari-bugfix","meta_rev_id":"8d134fb9583a2a97c3c536211a79615872a4a92c","_number":867174,"virtual_id_number":867174,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-12 13:30:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2022-12-12 09:51:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2022-12-12 13:23:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2022-12-12 13:23:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2022-12-10 18:05:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-10 18:05:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-12 09:51:16.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2022-12-12 13:23:26.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"f7f3d28a2035b86c779f9c120f0df448a3cdc0b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-12-10 18:04:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"907d4daef44b6d462126d0ea3027a6552dca0fb6","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-10 18:05:02.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"62ea66b28d0a55f977f3ed73242a664e5d1a92ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-10 18:11:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb736223c49e42dda2822ad233dac1f0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34352597ab374a1ba4ea4c69a77f597e : SUCCESS in 6m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6baf97c7176f4ffc82f904379f406963 : SUCCESS in 6m 22s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6a67f60b56834a93971ed623f8d7d59f : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"75e1ac7abed3d517cba768ed99d0d95d24be98bc","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-12-12 09:51:16.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b19016a4a8176205146d2fdd9bd7d95d80d48a82","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-12-12 13:23:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d24ca6e6b7d94b2c624827a33cc9f8ca8e2ae56a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 13:23:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c067c0622f5fbec8f47c878dc7af6fd14a16beb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 13:30:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd43c6710922429f9d5c68edfd60e719\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7be927a2b0c45e18e4a7c653d763408 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8283d52c22a4c92bfb3037a85f87af1 : SUCCESS in 6m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4515ec29792643548b9098d568ace788 : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":1},{"id":"8d134fb9583a2a97c3c536211a79615872a4a92c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 13:30:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a3bff2dbb8dc6fcd30f45c1f0c7987343fed28b6","revisions":{"a3bff2dbb8dc6fcd30f45c1f0c7987343fed28b6":{"kind":"REWORK","_number":1,"created":"2022-12-10 18:04:10.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/74/867174/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/74/867174/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/867174/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/867174/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/867174/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/74/867174/1"}}},"commit":{"parents":[{"commit":"b020ba6123d4692850ce1faf51658a5e2da56152","subject":"Merge \"Tag monthly kolla stable releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b020ba6123d4692850ce1faf51658a5e2da56152"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-12-10 18:03:39.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-12-10 18:03:39.000000000","tz":60},"subject":"[masakari] Release bugfix","message":"[masakari] Release bugfix\n\nChange-Id: I871156f0e8e5ebb36bb695ffb9b10c491b355337\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a3bff2dbb8dc6fcd30f45c1f0c7987343fed28b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a3bff2dbb8dc6fcd30f45c1f0c7987343fed28b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[]}
