)]}'
{"id":"openstack%2Freleases~868588","triplet_id":"openstack%2Freleases~master~I44a8c6fbecd2838ccb67994d9ded43891d3ad4e2","project":"openstack/releases","branch":"master","topic":"wallaby-last","attention_set":{},"removed_from_attention_set":{"13940":{"account":{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},"last_update":"2023-02-03 16:08:40.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-02-03 16:00:29.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"21387":{"account":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"last_update":"2022-12-27 00:29:52.000000000","reason":"\u003cGERRIT_ACCOUNT_21387\u003e replied on the change","reason_account":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-02-03 16:08:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I44a8c6fbecd2838ccb67994d9ded43891d3ad4e2","subject":"[freezer-tempest-plugin] Tag wallaby-last","status":"MERGED","created":"2022-12-26 04:57:43.000000000","updated":"2023-02-06 00:58:20.000000000","submitted":"2023-02-03 16:08:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868588-wallaby-last","meta_rev_id":"86dcb6dba2250ba87c2bf00c7e02abbbde8d5e6f","_number":868588,"virtual_id_number":868588,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-03 16:08:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2023-02-03 16:00:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2023-02-03 10:52:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2023-02-06 00:58:20.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2023-02-03 16:00:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"}],"reviewers":{"REVIEWER":[{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-26 04:58:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-26 05:04:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-26 12:20:39.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"state":"REVIEWER"},{"updated":"2022-12-26 12:20:39.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":13940,"name":"Saad Zaher","email":"eng.szaher@gmail.com","username":"szaher"},"state":"REVIEWER"},{"updated":"2023-01-04 09:45:27.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-01-04 14:32:50.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"a55efdc20ea6981c93762e9af0ac2228707c3e73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-26 04:57:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"098e1d96a29dcf5237eca8f4e3367474f2e99ce5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-26 05:04:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f382a70a4a5d4e729aed998c5faebcce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/772a717bb24b48629d38a44c7be66fa5 : SUCCESS in 5m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/326d6169a74d4b16b532f3c8e2a2886a : SUCCESS in 3m 57s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8405f1763c304cb49bc39214c3c59ede : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":1},{"id":"d59c601438061d1a7f7c9a3ff00ad48f2af77f28","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2022-12-27 00:29:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5fef9cf1dc159ba52bf1aab02f4b6b022b4864e4","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-27 00:30:34.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"1a78cbbae12aa6920f15ea22f79dce1909a32b5e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-01-04 09:45:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b845e6d7537fc72b5f4194540c8c92ec5d0da417","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-01-04 14:32:50.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a3bd57eb9ca4202e3f717894940f9a05b552a75f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-01-09 09:00:22.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"cba0eb55195cecaa9b2ecfd72ad11d1baddcca7a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-25 21:38:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a7010d8e7e4c7e1dca0fde1d10f64119401dcefc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-25 21:39:26.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"2f5aa27b5c67a4790875ca4c60fe8aedd8218f31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 21:48:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a850e820adab43b49b5ee40aa5a5968b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef8a4be3f9ee4bcaaab551bd90b9904d : SUCCESS in 7m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a36a3465f93847948fcd1640f1b8b028 : SUCCESS in 4m 40s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a03abfb905bf40019aa35c626514bc6e : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":2},{"id":"ebe22970fa36ff2ccceba11aa0c45f003010346e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-03 10:52:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"43687a083ef0a9ace4598373f3ce622c3792ac00","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-03 16:00:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fbb221d9e73ebee71a5c4b129c28ab042d0f1cb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 16:00:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aacadf045cf42baa163c51b4bdf13d829ef131b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 16:08:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e957be16fd56459aa894f57c1cdfb38d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3be94d3ededc422d849803a33836b3bc : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41d238d0e4db4f6cb9c4d8236c2f2bd9 : SUCCESS in 6m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0249f38d3ee642e6b514627f29a2776d : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":2},{"id":"25b43526004534c0a9795944701ccaa47e9dba6d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 16:08:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"86dcb6dba2250ba87c2bf00c7e02abbbde8d5e6f","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2023-02-06 00:58:20.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fc8e2c4f6f910cdd9f916c17c604467a9fcc8fb1","revisions":{"2768442e45f514d94baee1f359f593954628c557":{"kind":"REWORK","_number":1,"created":"2022-12-26 04:57:43.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/88/868588/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/88/868588/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/868588/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/868588/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/868588/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/88/868588/1"}}},"commit":{"parents":[{"commit":"c701c11b9d5fdd5ba97bb0f06e7c866c3d68b977","subject":"Merge \"Release python-cinderclient for Antelope-1 milestone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c701c11b9d5fdd5ba97bb0f06e7c866c3d68b977"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-26 04:57:17.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-26 04:57:17.000000000","tz":-360},"subject":"[freezer-tempest-plugin] Tag wallaby-last","message":"[freezer-tempest-plugin] Tag wallaby-last\n\nWallaby branch is in Extended Maintenance now[1]. Tempest and\ntempest plugins are branchless which means master version of\nTempest and its plugins is used to test the supported stable\nbranches.\n\nOnce any stable branch is moved to EM state then, Tempest and\nits plugins compatible tag[2] needs to be released so that we can\nkeep testing the EM stable branches with that tag once master\nTempest and its plugins are not compatible[3].\n\nTagging \u0027wallaby-last\u0027 as well as a new version also with same hash.\n\n[1] https://releases.openstack.org/wallaby/index.html\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: I44a8c6fbecd2838ccb67994d9ded43891d3ad4e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2768442e45f514d94baee1f359f593954628c557"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2768442e45f514d94baee1f359f593954628c557"}]},"branch":"refs/heads/master"},"fc8e2c4f6f910cdd9f916c17c604467a9fcc8fb1":{"kind":"REWORK","_number":2,"created":"2023-01-25 21:39:26.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/88/868588/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/88/868588/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/868588/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/868588/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/868588/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/88/868588/2"}}},"commit":{"parents":[{"commit":"c701c11b9d5fdd5ba97bb0f06e7c866c3d68b977","subject":"Merge \"Release python-cinderclient for Antelope-1 milestone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c701c11b9d5fdd5ba97bb0f06e7c866c3d68b977"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-26 04:57:17.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-01-25 21:39:25.000000000","tz":0},"subject":"[freezer-tempest-plugin] Tag wallaby-last","message":"[freezer-tempest-plugin] Tag wallaby-last\n\nWallaby branch is in Extended Maintenance now[1]. Tempest and\ntempest plugins are branchless which means master version of\nTempest and its plugins is used to test the supported stable\nbranches.\n\nOnce any stable branch is moved to EM state then, Tempest and\nits plugins compatible tag[2] needs to be released so that we can\nkeep testing the EM stable branches with that tag once master\nTempest and its plugins are not compatible[3].\n\nTagging \u0027wallaby-last\u0027 as well as a new version also with same hash.\n\n[1] https://releases.openstack.org/wallaby/index.html\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: I44a8c6fbecd2838ccb67994d9ded43891d3ad4e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fc8e2c4f6f910cdd9f916c17c604467a9fcc8fb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fc8e2c4f6f910cdd9f916c17c604467a9fcc8fb1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[]}
