)]}'
{"id":"openstack%2Freleases~870035","triplet_id":"openstack%2Freleases~master~Iab885d4e36f64903b323e16c74d8315c759584de","project":"openstack/releases","branch":"master","topic":"redirect-fix","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-01-19 09:47:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iab885d4e36f64903b323e16c74d8315c759584de","subject":"Fix upper-constraints redirect for new release naming","status":"MERGED","created":"2023-01-12 22:53:57.000000000","updated":"2023-01-19 09:47:55.000000000","submitted":"2023-01-19 09:47:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870035-redirect-fix","meta_rev_id":"08cce39212ec696ea45adaf2ee3b03a6197449ae","_number":870035,"virtual_id_number":870035,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-19 09:47:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2023-01-19 09:18:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2023-01-16 10:39:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2023-01-19 09:18:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-12 23:08:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-16 10:39:48.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-01-19 09:18:01.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"b07d9e9c9107ab186692e875ec2864b2e11ec48e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-01-12 22:53:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58747c8344ff7d6eff490fc78b14ee52f03b3a9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 23:08:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d260f903663645dc8ee7054d0ed85661\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdc50fafb397453db09983addcbdc1d1 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af48866f358844b79e5ac9b18fa09d6e : FAILURE in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f2e0cef7d0dc4d1c8ffa0d475c485a69 : FAILURE in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c3e7cab2bc894d0794dfe184e0a5c540 : FAILURE in 3m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0647aaf01798447c9b713876719e6409 : SUCCESS in 7m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/326c0c0524f44c759ed1a6e41b127a63 : FAILURE in 3m 53s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/473f5f49caff4a9380b44975540619f4 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"4074a62d2c24af79e2698b70b1a08f6f9cd51509","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-01-13 09:42:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4bcef1444efa52424c1b978996e254792ad7f179","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 09:50:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/161f7423d1864b6695e8e509ee4a8993\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2d3ad447aba463b80f66df90a90e0e0 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/487560fe303047c78b08c88cbe3ce2f2 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5907b1888b8c449fac37184a3e5c7aed : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/22d69d6655cf4d4aa41ea569dc449df2 : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef2c177ed0664544b074c872d6a4a773 : SUCCESS in 6m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/69a217222b084f41bd45f496b677de3b : FAILURE in 4m 02s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e43197ad6e7d46db9ede30ff2a2fc48f : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":2},{"id":"8c96e72497b07b4317dc40e76c4ad153f1ebbe47","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-01-13 09:58:23.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d0701cb0b893654d8fecc187295cffebac92d072","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-01-14 08:40:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"88e24c70705cc27bdb573fdbe0fa3a834b784bf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-14 08:52:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2eda6ba071964e60829e1bafdfbddd60\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38a42edeb4554fffa9f1bbf9e1ffc289 : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/546f7282715045819ab717adfab41399 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a97c48c5e7a74d779a706957dad60cc5 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c824fc9ec10442399f6030ac6e2918e : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb3f7641ac4846438620b155bacf4843 : SUCCESS in 5m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e1079bb48cf444fcab85636259bef897 : SUCCESS in 4m 20s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/37bbf5d57608498b81cf206ecd1a63cb : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":3},{"id":"1ae20f9b62d7eef34270c4d954a425ceda013794","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-01-14 09:10:52.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2fbeae7b49990f50c71702edd36219268f5e7fdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-01-14 09:30:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ad0e6750fe66197747e4402b7ee1afe068c65ed6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-14 09:45:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16d9e4b09e454087aa28c70d0d3252cf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ec14de3ce084559b8a744173b09c3bf : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90bf2e8f33d94278a9e410daeb729e1f : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2541b8f7aeb422792bd442353647cbc : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a90ac6e3983b4da1a5f96a7e8c294df2 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b7f37c1a6c347708aa8f6f9fdae42b8 : SUCCESS in 7m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b74f24affb0a47d58e57c3afa4d8dd9e : SUCCESS in 5m 10s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8fff8e2b332b4b5297e372cbd38c4238 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":4},{"id":"1fc47078e8a00bba02c2a3e33c0f05956b05d6df","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-01-16 10:39:48.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6d880cc4a4ce3e2223a28b073e47992e447f3b25","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-01-19 09:18:01.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1063f0ab9ccf03e9637027ee455a586529786a3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 09:18:40.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"367e9fdea834ef9220ac3f5da00f38ae8131b014","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 09:47:55.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c98899b7e75d490a83bdde76ebf1516c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f96a28f85c4492d9c9940b0cf73a41b : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5633535a49b9411b9ca21b8c84267930 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/089835ede64b44ee98e82d6450643b12 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/545e98ef47aa414d8e6c389e32f98236 : SUCCESS in 6m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e3763a5c5ced497e870ed976f510b19f : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1ad64890f12249d4a18c8e50daa1602f : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":4},{"id":"08cce39212ec696ea45adaf2ee3b03a6197449ae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 09:47:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2b8ee3aaeca32e49d54fc85e425c304984e4e5dd","revisions":{"7b429b5f844e71cd53f060aff1d0ce2b513d8dd9":{"kind":"REWORK","_number":1,"created":"2023-01-12 22:53:57.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/35/870035/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/35/870035/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/870035/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/870035/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/870035/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/35/870035/1"}}},"commit":{"parents":[{"commit":"7bdcec84c30232fd612bfb07b7830a87bb332be5","subject":"Merge \"[zaqar-tempest-plugin] Tag wallaby-last\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7bdcec84c30232fd612bfb07b7830a87bb332be5"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-01-12 22:36:39.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-01-12 22:45:51.000000000","tz":60},"subject":"WIP: Fix redirect for new naming","message":"WIP: Fix redirect for new naming\n\nChange-Id: Iab885d4e36f64903b323e16c74d8315c759584de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7b429b5f844e71cd53f060aff1d0ce2b513d8dd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7b429b5f844e71cd53f060aff1d0ce2b513d8dd9"}]},"branch":"refs/heads/master"},"b4f20c484950cc18868cb5aac31ca6f422f6321e":{"kind":"REWORK","_number":2,"created":"2023-01-13 09:42:16.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/35/870035/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/35/870035/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/870035/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/870035/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/870035/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/35/870035/2"}}},"commit":{"parents":[{"commit":"7bdcec84c30232fd612bfb07b7830a87bb332be5","subject":"Merge \"[zaqar-tempest-plugin] Tag wallaby-last\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7bdcec84c30232fd612bfb07b7830a87bb332be5"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-01-12 22:36:39.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-01-13 09:40:22.000000000","tz":60},"subject":"Fix redirect for new release naming","message":"Fix redirect for new release naming\n\nThe new release identification / naming schema [1] (like:\n2023.1 Antelope) broke the redirection rules for branch specific upper\nconstraint files as the stable branch names are from now on based on\nthe new release identification (like: stable/2023.1).\n\nThis patch fixes the redirections by introducing the release ID field\nto series status, to translate a release name to \u0027stable/\u003crelease_id\u003e\u0027\nwhere this ID exists.\n\n[1] https://governance.openstack.org/tc/reference/release-naming.html\n\nChange-Id: Iab885d4e36f64903b323e16c74d8315c759584de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b4f20c484950cc18868cb5aac31ca6f422f6321e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b4f20c484950cc18868cb5aac31ca6f422f6321e"}]},"branch":"refs/heads/master"},"9372a2d45f5c6acd7241859569f5f8a78d18f1e5":{"kind":"REWORK","_number":3,"created":"2023-01-14 08:40:29.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/35/870035/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/35/870035/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/870035/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/870035/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/870035/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/35/870035/3"}}},"commit":{"parents":[{"commit":"7bdcec84c30232fd612bfb07b7830a87bb332be5","subject":"Merge \"[zaqar-tempest-plugin] Tag wallaby-last\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7bdcec84c30232fd612bfb07b7830a87bb332be5"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-01-12 22:36:39.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-01-14 08:39:42.000000000","tz":60},"subject":"Fix upper-constraints redirect for new release naming","message":"Fix upper-constraints redirect for new release naming\n\nThe new release identification / naming schema [1] (like:\n2023.1 Antelope) broke the redirection rules for branch specific upper\nconstraint files as the stable branch names are from now on based on\nthe new release identification (like: stable/2023.1).\n\nThis patch fixes the redirections by introducing the release ID field\nto series status, to translate a release name to \u0027stable/\u003crelease_id\u003e\u0027\nwhere this ID exists.\n\n[1] https://governance.openstack.org/tc/reference/release-naming.html\n\nChange-Id: Iab885d4e36f64903b323e16c74d8315c759584de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9372a2d45f5c6acd7241859569f5f8a78d18f1e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9372a2d45f5c6acd7241859569f5f8a78d18f1e5"}]},"branch":"refs/heads/master"},"2b8ee3aaeca32e49d54fc85e425c304984e4e5dd":{"kind":"REWORK","_number":4,"created":"2023-01-14 09:30:19.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/35/870035/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/35/870035/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/870035/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/870035/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/870035/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/35/870035/4"}}},"commit":{"parents":[{"commit":"7bdcec84c30232fd612bfb07b7830a87bb332be5","subject":"Merge \"[zaqar-tempest-plugin] Tag wallaby-last\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7bdcec84c30232fd612bfb07b7830a87bb332be5"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-01-12 22:36:39.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-01-14 09:29:27.000000000","tz":60},"subject":"Fix upper-constraints redirect for new release naming","message":"Fix upper-constraints redirect for new release naming\n\nThe new release identification / naming schema [1] (like:\n2023.1 Antelope) broke the redirection rules for branch specific upper\nconstraint files as the stable branch names are from now on based on\nthe new release identification (like: stable/2023.1).\n\nThis patch fixes the redirections by introducing the release ID field\nto series status, to translate a release name to \u0027stable/\u003crelease-id\u003e\u0027\nwhere this ID exists.\n\n[1] https://governance.openstack.org/tc/reference/release-naming.html\n\nChange-Id: Iab885d4e36f64903b323e16c74d8315c759584de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2b8ee3aaeca32e49d54fc85e425c304984e4e5dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2b8ee3aaeca32e49d54fc85e425c304984e4e5dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[]}
