)]}'
{"id":"openstack%2Freleases~870593","triplet_id":"openstack%2Freleases~master~I874985268f083b38e39f584e3fa8e440e01fe1b4","project":"openstack/releases","branch":"master","topic":"xena-stable","attention_set":{},"removed_from_attention_set":{"16465":{"account":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"last_update":"2023-01-18 12:38:34.000000000","reason":"Change was abandoned"},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2023-01-17 13:39:57.000000000","reason":"\u003cGERRIT_ACCOUNT_7414\u003e replied on the change","reason_account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-01-18 12:38:34.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I874985268f083b38e39f584e3fa8e440e01fe1b4","subject":"Release keystonemiddleware for stable/xena","status":"ABANDONED","created":"2023-01-16 15:43:05.000000000","updated":"2023-01-18 12:38:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3d0cfc655b6d873f35683c764e22b2ca1de69bff","_number":870593,"virtual_id_number":870593,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-01-16 16:13:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2023-01-17 13:39:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2023-01-17 13:40:50.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-16 15:43:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-16 16:13:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-16 16:20:57.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2023-01-16 16:20:58.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"}],"messages":[{"id":"6f7614090dc8b067c59806f982759bb9b54369ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-01-16 15:43:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc8d63ac62c7b327f06866bba9edc9b9713d99eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 16:13:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dc8eb661f714431add275ef6d8fcb58\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c37b32064164d94835a6fd937b65acb : SUCCESS in 6m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9395045bf6304c36a9e159db0b03dcd9 : SUCCESS in 5m 46s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/80520071212840ff8e30a0b0f58ce981 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"4e6ce7be4e7e7d5cd2ed4a18c863fb6d841c2f38","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-01-17 13:39:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"147551d4f5f34f3705da8b5c7179d8cd2ad53b80","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 13:40:50.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"3d0cfc655b6d873f35683c764e22b2ca1de69bff","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-01-18 12:38:34.000000000","message":"Abandoned\n\nNo significant changes landed, aborting","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a81e0831bdde886c361c41e72a9ba26f76770272","revisions":{"a81e0831bdde886c361c41e72a9ba26f76770272":{"kind":"REWORK","_number":1,"created":"2023-01-16 15:43:05.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/93/870593/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/93/870593/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/870593/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/870593/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/870593/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/93/870593/1"}}},"commit":{"parents":[{"commit":"8cdd1f970520eec4a50a6cf91028aff7bc49327c","subject":"Merge \"[telemetry-tempest-plugin] Tag wallaby-last\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8cdd1f970520eec4a50a6cf91028aff7bc49327c"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-01-16 15:43:01.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-01-16 15:43:01.000000000","tz":60},"subject":"Release keystonemiddleware for stable/xena","message":"Release keystonemiddleware for stable/xena\n\nThis release picks up new commits to keystonemiddleware since\nthe last release from stable/xena.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\n$ git log --oneline --no-merges 9.3.0..5c318d6\n5c318d6 Xena: Update classifiers according to tested runtimes\n200f738 Xena: Update Python 3 job template\n5d907d2 Update TOX_CONSTRAINTS_FILE for stable/xena\n711c9e7 Update .gitreview for stable/xena\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I874985268f083b38e39f584e3fa8e440e01fe1b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a81e0831bdde886c361c41e72a9ba26f76770272"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a81e0831bdde886c361c41e72a9ba26f76770272"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
