)]}'
{"id":"openstack%2Freleases~871512","triplet_id":"openstack%2Freleases~master~I0e164e15a5157f6e6a42ef7ca154f207308641b7","project":"openstack/releases","branch":"master","topic":"eol-delete-speedup","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-02-03 11:09:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0e164e15a5157f6e6a42ef7ca154f207308641b7","subject":"EOL branch delete script speed up","status":"MERGED","created":"2023-01-23 16:27:20.000000000","updated":"2023-02-03 11:09:31.000000000","submitted":"2023-02-03 11:09:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871512-eol-delete-speedup","meta_rev_id":"56ac428bcfd70a92fc43ddf470d0edc7fc2173f4","_number":871512,"virtual_id_number":871512,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-03 11:09:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2023-01-26 12:32:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2023-02-03 10:49:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2023-02-03 10:49:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-23 16:45:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-26 12:32:42.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2023-02-03 10:49:40.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"0bc84645a32236113f1d787f10ff2bb623e0d68c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-01-23 16:27:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e33a5c82336efcf0a05516db74c1f84eee797d30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 16:45:50.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b1f1df340374aa594a5824bea91d9fa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d25a3b830084911ae3d5464f1c758ed : FAILURE in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5c79dd58cf84076b5a66ab2faaf2de6 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fdd672a874f948a899a6451c6c88e79b : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b75c884ebccc441aa29c89e29344ba3a : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4be5198996b7452db8ce5920bdfc2c01 : SUCCESS in 7m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ad25028bdc654377bf5ab847a5bb4cb7 : SUCCESS in 5m 29s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2cc018327a774301b4c1a2e880d6a4b6 : SUCCESS in 3m 45s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/4d6c0200ff434a7e9b221c9ec5343311 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"f7ef900ade7de4bdd4fd44651c70ec30f2e6445f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-01-25 19:47:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"501472e694af726904450dcab5908d0cfebf7959","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 19:59:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c837bd1da37040c4beff3448de1f085f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4910f841fe8458bafbf597dde3a5496 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc26daee0c094c4ca6cbc70cb33d6124 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54651a59823d412aabd9ec02b8c24d59 : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba20ebc52d99446f81042a67f022e391 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d654fd3adad74e77b35dcf9bd7bf4796 : SUCCESS in 6m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5a8f78d517c246f882275fead53a2374 : SUCCESS in 4m 51s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e18758e984944681905af62826b93d05 : SUCCESS in 3m 25s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/c68799341f1b4b49a3f7ce949cbd3e81 : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":2},{"id":"ee16a7f73679e9c9c96520651dd17cd8c535a9b8","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-01-26 12:32:42.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e9e28ae0dcdde8b663fd2cf6f1b87992ca7c71fc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-03 10:49:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b005d0ea97253434aa9c1f56e2c6cfb5363b9b2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 10:50:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0e135a4354d3757af25ea0b8fa5eaee3940601d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 11:09:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e482ee1fc73c43debe5ebe7bf73469df\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47d3d1ee9e164905b3435525cb0059d2 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/078521a2d83042ad91386b0000e08681 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66783d8fdc6244e7b37745bf932cf0bc : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fc4735c3a6445319fc579f4eb2c3bbc : SUCCESS in 7m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bf971c5015944e1087c2bdcc84bbc7f1 : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b25a7fb6620403189d71589beb4acb9 : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":2},{"id":"56ac428bcfd70a92fc43ddf470d0edc7fc2173f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 11:09:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c73a4dd0d7bd90f82c9a32173f83e64ac0df30e1","revisions":{"ee61ee57919d7f7b285e9e574b526c3a75798532":{"kind":"REWORK","_number":1,"created":"2023-01-23 16:27:20.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/12/871512/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/12/871512/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/871512/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/871512/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/871512/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/12/871512/1"}}},"commit":{"parents":[{"commit":"1502a9d517f5acbfe74c0f2473ff31014cd33c34","subject":"Add progress info printouts to list_eol_stale_branches.sh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1502a9d517f5acbfe74c0f2473ff31014cd33c34"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-12-05 18:23:24.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-01-23 16:13:26.000000000","tz":60},"subject":"EOL branch delete script speed up","message":"EOL branch delete script speed up\n\nThe script originally clones every repository and branch, which takes\nlong time. This patch queries GITEA instead whether the branch exists\nthere and only clones a repository if it is necessary.\n\nChange-Id: I0e164e15a5157f6e6a42ef7ca154f207308641b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ee61ee57919d7f7b285e9e574b526c3a75798532"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ee61ee57919d7f7b285e9e574b526c3a75798532"}]},"branch":"refs/heads/master"},"c73a4dd0d7bd90f82c9a32173f83e64ac0df30e1":{"kind":"REWORK","_number":2,"created":"2023-01-25 19:47:19.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/12/871512/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/12/871512/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/871512/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/871512/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/12/871512/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/12/871512/2"}}},"commit":{"parents":[{"commit":"1502a9d517f5acbfe74c0f2473ff31014cd33c34","subject":"Add progress info printouts to list_eol_stale_branches.sh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1502a9d517f5acbfe74c0f2473ff31014cd33c34"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2022-12-05 18:23:24.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2023-01-25 19:25:04.000000000","tz":60},"subject":"EOL branch delete script speed up","message":"EOL branch delete script speed up\n\nThe script originally clones every repository and branch, which takes\nlong time. This patch queries GITEA instead whether the branch exists\nthere and only clones a repository if it is necessary.\n\nChange-Id: I0e164e15a5157f6e6a42ef7ca154f207308641b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c73a4dd0d7bd90f82c9a32173f83e64ac0df30e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c73a4dd0d7bd90f82c9a32173f83e64ac0df30e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY"}]}],"submit_requirements":[]}
