)]}'
{"id":"openstack%2Freleases~873523","triplet_id":"openstack%2Freleases~master~If3c7e9428f3c2e3531302b51eb8ca19821ba7da4","project":"openstack/releases","branch":"master","topic":"antelope-milestone-3","attention_set":{},"removed_from_attention_set":{"28356":{"account":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"last_update":"2023-02-13 12:28:55.000000000","reason":"\u003cGERRIT_ACCOUNT_28356\u003e replied on the change","reason_account":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-02-17 16:43:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If3c7e9428f3c2e3531302b51eb8ca19821ba7da4","subject":"Release final python-cloudkittyclient for Antelope","status":"MERGED","created":"2023-02-13 10:03:45.000000000","updated":"2023-02-17 16:43:18.000000000","submitted":"2023-02-17 16:43:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873523-antelope-milestone-3","meta_rev_id":"e20a4879938af6f53794693855beb031eb483b11","_number":873523,"virtual_id_number":873523,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-17 16:43:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"all":[{"value":1,"date":"2023-02-17 11:45:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},{"value":2,"date":"2023-02-17 12:27:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},{"value":1,"date":"2023-02-17 16:23:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2023-02-17 11:46:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-13 10:04:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-13 10:44:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-13 12:21:32.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"state":"REVIEWER"},{"updated":"2023-02-17 12:27:24.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"58a816b7b666c2c93c3b767ba3fdae3bafb6d592","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-13 10:03:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f767a631726e5aacf5c2212f0a5f16e16f1551fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 10:44:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ae784cc4e434c079c50d803c0bd6f5e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a38af07bedf04d6cabb75c58e89a3e21 : SUCCESS in 6m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ca37840fd08f4c329a147850ccff1921 : SUCCESS in 4m 50s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/33a3f69a94dc4e5da582799344a41a9f : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":1},{"id":"677192a494c8e3826408703a077e6c977ec79846","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-02-13 12:28:55.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6188c22cd9db45680f6d63364ac8560a5588fbe2","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 12:29:46.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"fb0165b169cb6798435900214754b547adbdcebe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-13 13:43:31.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e7e923813a55dcd3b5d268bb61af083bac196f36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 13:56:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f3a2d3046fd4fe292797d27af5fb253\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9070446c002e49dea402133697178f94 : SUCCESS in 6m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cecbe8491c724eb29095c3dbc464238c : SUCCESS in 6m 23s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0eda50466fd84df8bfc00c2ece99acd4 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":2},{"id":"dd99eb3eaf91f6c9ab3f8b04782573e526a3bbe3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-13 13:57:02.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"255c88f4e7ce01e9491845adeb44b84714ce02f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 14:08:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c819a478bb4944b6821530cfe94cd50d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/963090c1ff90476d837788d18c7ed553 : SUCCESS in 7m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/96d9224ab04940f38f58111a89d9c62c : SUCCESS in 5m 36s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0089a6332f864dbcbdd95ece6b71f95b : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":3},{"id":"94e0d07414ea31bf38cab5efd69ce362201856d1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-17 10:14:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"52aa252ca93c996ba7099280d6d9aff91dc290c9","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-02-17 11:45:04.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"979c6af7b7096750a0c082f02262bdd0ddb60de5","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 11:46:47.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"c5da71c675932a8fa967fecb09879d3cee46ea84","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-17 12:27:24.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"de3b491f3f46c14dd03721b42788fd88151a3012","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-17 16:23:58.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cd9ee4247838d2df41201b4d25eb48b740337b19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 16:24:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"789b4afac76c1855b9a8966cfd045005efaf9d7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 16:43:18.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38b01a32e87e4b4f874181d3fe7a899a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d25a4a8550d4ea38e4fd9fff46e3f35 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad84cece3ad447efada9575422fd87cd : SUCCESS in 7m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c9f4ba0a3a9d4209a8ba7e1546c68562 : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":3},{"id":"e20a4879938af6f53794693855beb031eb483b11","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 16:43:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"28200dcbd0b8fb71cd534503d2ec9632eddd6880","revisions":{"9a13f5c7d43e8ca5693788d333a9fefe9e064393":{"kind":"REWORK","_number":1,"created":"2023-02-13 10:03:45.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/23/873523/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/23/873523/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/873523/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/873523/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/873523/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/23/873523/1"}}},"commit":{"parents":[{"commit":"68fd755f8bb76712157186562ab3168d968c89ff","subject":"Merge \"Release final stevedore for 2023.1 Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/68fd755f8bb76712157186562ab3168d968c89ff"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-13 10:03:41.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-13 10:03:41.000000000","tz":60},"subject":"Release final python-cloudkittyclient for Zed","message":"Release final python-cloudkittyclient for Zed\n\nThis is a release for python-cloudkittyclient to get any merged but\nunreleased changes out prior to the client lib freeze for Zed going\ninto effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Friday, February 16th.\n\n$ git log --oneline --no-merges 4.6.0..97dc4d6\n97dc4d6 Switch to 2023.1 Python3 unit tests and generic template name\n0c97e40 Update master for stable/zed\n0ee5e17 remove unicode prefix from code\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: If3c7e9428f3c2e3531302b51eb8ca19821ba7da4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9a13f5c7d43e8ca5693788d333a9fefe9e064393"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9a13f5c7d43e8ca5693788d333a9fefe9e064393"}]},"branch":"refs/heads/master"},"31fe0e831a4c3e83e1cc6405238c302efa125277":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-02-13 13:43:31.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/23/873523/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/23/873523/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/873523/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/873523/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/873523/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/23/873523/2"}}},"commit":{"parents":[{"commit":"68fd755f8bb76712157186562ab3168d968c89ff","subject":"Merge \"Release final stevedore for 2023.1 Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/68fd755f8bb76712157186562ab3168d968c89ff"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-13 10:03:41.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2023-02-13 13:43:31.000000000","tz":0},"subject":"Release final python-cloudkittyclient for Zed","message":"Release final python-cloudkittyclient for Zed\n\nThis is a release for python-cloudkittyclient to get any merged but\nunreleased changes out prior to the client lib freeze for Zed going\ninto effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Friday, February 17th.\n\n$ git log --oneline --no-merges 4.6.0..97dc4d6\n97dc4d6 Switch to 2023.1 Python3 unit tests and generic template name\n0c97e40 Update master for stable/zed\n0ee5e17 remove unicode prefix from code\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: If3c7e9428f3c2e3531302b51eb8ca19821ba7da4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/31fe0e831a4c3e83e1cc6405238c302efa125277"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/31fe0e831a4c3e83e1cc6405238c302efa125277"}]},"branch":"refs/heads/master","description":"Edit commit message"},"28200dcbd0b8fb71cd534503d2ec9632eddd6880":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-02-13 13:57:02.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/23/873523/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/23/873523/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/873523/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/873523/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/873523/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/23/873523/3"}}},"commit":{"parents":[{"commit":"68fd755f8bb76712157186562ab3168d968c89ff","subject":"Merge \"Release final stevedore for 2023.1 Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/68fd755f8bb76712157186562ab3168d968c89ff"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-13 10:03:41.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2023-02-13 13:57:02.000000000","tz":0},"subject":"Release final python-cloudkittyclient for Antelope","message":"Release final python-cloudkittyclient for Antelope\n\nThis is a release for python-cloudkittyclient to get any merged but\nunreleased changes out prior to the client lib freeze for Antelope going\ninto effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Friday, February 17th.\n\n$ git log --oneline --no-merges 4.6.0..97dc4d6\n97dc4d6 Switch to 2023.1 Python3 unit tests and generic template name\n0c97e40 Update master for stable/zed\n0ee5e17 remove unicode prefix from code\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: If3c7e9428f3c2e3531302b51eb8ca19821ba7da4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/28200dcbd0b8fb71cd534503d2ec9632eddd6880"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/28200dcbd0b8fb71cd534503d2ec9632eddd6880"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[]}
