)]}'
{"id":"openstack%2Freleases~873526","triplet_id":"openstack%2Freleases~master~I384606df6c44e84a62f941fd5d75c40a75c159c3","project":"openstack/releases","branch":"master","topic":"antelope-milestone-3","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2023-02-17 10:17:46.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-02-17 09:58:36.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}},"hashtags":[],"change_id":"I384606df6c44e84a62f941fd5d75c40a75c159c3","subject":"Release final python-glanceclient for Antelope","status":"MERGED","created":"2023-02-13 10:04:52.000000000","updated":"2023-02-17 10:17:46.000000000","submitted":"2023-02-17 10:17:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873526-antelope-milestone-3","meta_rev_id":"0648f467738a81a8c2512a55b65fc76feb573b03","_number":873526,"virtual_id_number":873526,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-17 10:17:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2023-02-16 13:38:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2023-02-17 09:58:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2023-02-17 09:58:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2023-02-16 12:12:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-13 10:05:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-13 10:19:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-13 12:22:02.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2023-02-16 13:38:49.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"7aee75cbca55eaf7aa74914eb83e1543933cb79e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-13 10:04:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1da40ba9d6faa4f2fd7d2b69adce1cb100745e60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 10:19:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a73d411da8341d4ad585acf7802867d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b21c50038cc64fd79f9aede2e9d65b09 : SUCCESS in 6m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/18b814ec17e64bbcbb74c5e643cc739e : SUCCESS in 5m 52s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4e539b5b3fae4defb9a3ade352c20dda : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"e14c05106fc1c488b0284100654457196236bdb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-13 13:44:32.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"09da3cff9bc190d7325276576e85b26fd0a218c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-13 14:04:02.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"293a3464ff339deb148f11aefba6bc8323070f72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 14:17:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8b0486f256f458ebf28fa6f7c15adec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e847f1d06b44627880e78aa27198181 : SUCCESS in 6m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ff0a690cf70e4fc4afa6c47979d83040 : SUCCESS in 5m 15s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/30d93d545ed64f4ea455f5d97a88bd41 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":3},{"id":"a2643be6c7d5e0020ff93968eb9a696abd51ffba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-16 12:11:32.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"9b364eca027a6c9aa2d3fe439d821c03ac44685a","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 12:12:30.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"cdd2a8abf95ecf0c181471d5f5256cb5a32d35e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 12:35:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/124ac7c17c6d403d85f9002b854bda14\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86924b54622c47dcb5327e5c51f1af6b : SUCCESS in 6m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ba0559c54ca442baba9a892a6838df7f : SUCCESS in 6m 23s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c98c2ae1433342e0a93216f31cc57714 : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":4},{"id":"aba02ce141aff8f693787b0a5abc51ea37585852","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-16 13:38:49.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"91280c6627f062cbae90111d6b7ddfb42fd83cd6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-17 09:58:36.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"918fab18a423c9b63720726df3afeecf9766aa47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 09:59:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"fe152f227c7724b072b66e80ec377e4fd6daada5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 10:17:46.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a73f165c57df42cb9ca67ada06c48cb0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62e1cb33f4864edcbaad56969ca4639a : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5a15d068deb4923b3d067d35cd6ba26 : SUCCESS in 6m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f22ef295100d4970a1deac52d486c6c9 : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":4},{"id":"0648f467738a81a8c2512a55b65fc76feb573b03","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 10:17:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"cf53abe3b59efb4751a15b5e9244ae63b1963325","revisions":{"94a9dd84d7ab24502bad3787874c42b4c5160960":{"kind":"REWORK","_number":1,"created":"2023-02-13 10:04:52.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/26/873526/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/26/873526/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/873526/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/873526/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/873526/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/26/873526/1"}}},"commit":{"parents":[{"commit":"68fd755f8bb76712157186562ab3168d968c89ff","subject":"Merge \"Release final stevedore for 2023.1 Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/68fd755f8bb76712157186562ab3168d968c89ff"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-13 10:04:48.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-13 10:04:48.000000000","tz":60},"subject":"Release final python-glanceclient for Zed","message":"Release final python-glanceclient for Zed\n\nThis is a release for python-glanceclient to get any merged but\nunreleased changes out prior to the client lib freeze for Zed going\ninto effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Friday, February 16th.\n\n$ git log --oneline --no-merges 4.2.0..2d893b1\n6c95122 Fix functional tests and docs generation\n88e3b0a Boolean options: use strict checking\n4bb406d Unhardcode the value of DEFAULT_PAGE_SIZE from the tests\nfc8f9ac Remove unicode-related Python2-only code\nbbfe6cd Replace osc with glance commands\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I384606df6c44e84a62f941fd5d75c40a75c159c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/94a9dd84d7ab24502bad3787874c42b4c5160960"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/94a9dd84d7ab24502bad3787874c42b4c5160960"}]},"branch":"refs/heads/master"},"954fa684b0f08bc9a02a88b3577651ecab377e88":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-02-13 13:44:32.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/26/873526/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/26/873526/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/873526/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/873526/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/873526/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/26/873526/2"}}},"commit":{"parents":[{"commit":"68fd755f8bb76712157186562ab3168d968c89ff","subject":"Merge \"Release final stevedore for 2023.1 Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/68fd755f8bb76712157186562ab3168d968c89ff"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-13 10:04:48.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2023-02-13 13:44:32.000000000","tz":0},"subject":"Release final python-glanceclient for Zed","message":"Release final python-glanceclient for Zed\n\nThis is a release for python-glanceclient to get any merged but\nunreleased changes out prior to the client lib freeze for Zed going\ninto effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Friday, February 17th.\n\n$ git log --oneline --no-merges 4.2.0..2d893b1\n6c95122 Fix functional tests and docs generation\n88e3b0a Boolean options: use strict checking\n4bb406d Unhardcode the value of DEFAULT_PAGE_SIZE from the tests\nfc8f9ac Remove unicode-related Python2-only code\nbbfe6cd Replace osc with glance commands\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I384606df6c44e84a62f941fd5d75c40a75c159c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/954fa684b0f08bc9a02a88b3577651ecab377e88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/954fa684b0f08bc9a02a88b3577651ecab377e88"}]},"branch":"refs/heads/master","description":"Edit commit message"},"73f32bd9c753154246beb7e74e4f677ff926bb92":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-02-13 14:04:02.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/26/873526/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/26/873526/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/873526/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/873526/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/873526/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/26/873526/3"}}},"commit":{"parents":[{"commit":"68fd755f8bb76712157186562ab3168d968c89ff","subject":"Merge \"Release final stevedore for 2023.1 Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/68fd755f8bb76712157186562ab3168d968c89ff"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-13 10:04:48.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2023-02-13 14:04:02.000000000","tz":0},"subject":"Release final python-glanceclient for Antelope","message":"Release final python-glanceclient for Antelope\n\nThis is a release for python-glanceclient to get any merged but\nunreleased changes out prior to the client lib freeze for Antelope going\ninto effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Friday, February 17th.\n\n$ git log --oneline --no-merges 4.2.0..2d893b1\n6c95122 Fix functional tests and docs generation\n88e3b0a Boolean options: use strict checking\n4bb406d Unhardcode the value of DEFAULT_PAGE_SIZE from the tests\nfc8f9ac Remove unicode-related Python2-only code\nbbfe6cd Replace osc with glance commands\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I384606df6c44e84a62f941fd5d75c40a75c159c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/73f32bd9c753154246beb7e74e4f677ff926bb92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/73f32bd9c753154246beb7e74e4f677ff926bb92"}]},"branch":"refs/heads/master","description":"Edit commit message"},"cf53abe3b59efb4751a15b5e9244ae63b1963325":{"kind":"REWORK","_number":4,"created":"2023-02-16 12:11:32.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/26/873526/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/26/873526/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/873526/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/873526/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/26/873526/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/26/873526/4"}}},"commit":{"parents":[{"commit":"68fd755f8bb76712157186562ab3168d968c89ff","subject":"Merge \"Release final stevedore for 2023.1 Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/68fd755f8bb76712157186562ab3168d968c89ff"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-02-13 10:04:48.000000000","tz":60},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-02-16 12:11:32.000000000","tz":0},"subject":"Release final python-glanceclient for Antelope","message":"Release final python-glanceclient for Antelope\n\nThis is a release for python-glanceclient to get any merged but\nunreleased changes out prior to the client lib freeze for Antelope going\ninto effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Friday, February 17th.\n\n$ git log --oneline --no-merges 4.2.0..2d893b1\n6c95122 Fix functional tests and docs generation\n88e3b0a Boolean options: use strict checking\n4bb406d Unhardcode the value of DEFAULT_PAGE_SIZE from the tests\nfc8f9ac Remove unicode-related Python2-only code\nbbfe6cd Replace osc with glance commands\n\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I384606df6c44e84a62f941fd5d75c40a75c159c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cf53abe3b59efb4751a15b5e9244ae63b1963325"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cf53abe3b59efb4751a15b5e9244ae63b1963325"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[]}
